)]}'
{"id":"openstack%2Ftrove~677768","triplet_id":"openstack%2Ftrove~stable%2Fstein~I35e17afb9e827b1fead9d28dbf32f11ce4034a9b","project":"openstack/trove","branch":"stable/stein","topic":"fix-trove-ci","hashtags":[],"change_id":"I35e17afb9e827b1fead9d28dbf32f11ce4034a9b","subject":"Fix Trove CI jobs","status":"MERGED","created":"2019-08-21 15:42:39.000000000","updated":"2019-10-05 10:20:48.000000000","submitted":"2019-10-05 10:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677768-1570270760895-0f5a3271","meta_rev_id":"3188668dc64d03a1bbdaaa4aca93aef1d1a7c3d5","_number":677768,"virtual_id_number":677768,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 10:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-10-05 08:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-10-05 08:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 21:35:39.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-05 08:32:51.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-10-05 10:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"693f4cf092ec4cb3f727a5d81f3c3f2f3c5dbce5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 15:42:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e7a3447bf7ec2f82b538be21e1aaed66f3e16df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 15:45:35.000000000","message":"Topic set to fix-trove-ci","accounts_in_message":[],"_revision_number":1},{"id":"14774125549f7704a1ae467bd79b42e080eea10b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 16:31:10.000000000","message":"Patch Set 1:\n\nTrove\u0027s periodic stable jobs are failing [1], let\u0027s see if this patch would fix all the issues.\n\n[1] http://lists.openstack.org/pipermail/openstack-stable-maint/2019-August/thread.html#14214","accounts_in_message":[],"_revision_number":1},{"id":"3b694610bdf69d52e2efa5bf12de6cb3aa967185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 16:32:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d5045c965444cbbbc8cd13a812cd06 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1f9742ee4b4ca8ae675d8f85e77564 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f577823d18c0445a98c8a15e7878f339 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3931bcd2ef4c4554af7673391a89927f : SUCCESS in 10m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4e2a347ce9648aab7331063686df6a9 : SUCCESS in 17m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dca0332f69d3480694c5ac180827854d : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82a10b4858f4cb8a094fa1a43563e57 : SUCCESS in 5m 38s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d4fd899529b2402d881c5f17737c7777 : FAILURE in 44m 00s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/28f20d4ca7114226a6aaead0b638f06a : FAILURE in 38m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0a526d4682d142a7a7f5da91ed03c49e : FAILURE in 38m 40s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/025f121deb0040f99c6e30c9963d5ef3 : FAILURE in 33m 33s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/73f95ac2223e4f498303013b834fc16a : FAILURE in 41m 26s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4b47e985c9514eae9d333bb0e5254f0a : FAILURE in 38m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/727f33fcde3b47fd9e254df9e92ea0f4 : FAILURE in 42m 02s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/7d7edeb8494b45b89e574d2540c17499 : FAILURE in 42m 46s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/aca60cc68daf4dd095a1401695176fee : FAILURE in 41m 57s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9f3ee373d215451e91094f46b820e015 : FAILURE in 43m 28s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/827070fc76f443baa7d20e095cacb28a : FAILURE in 43m 03s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/5af0ac192b3843f8af413466c7952e60 : SUCCESS in 9m 15s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/2c75c0d1c8a54f50a56b58872ebb7356 : FAILURE in 5m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6448aba41ff41c897ae262dbcced3df : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0aa378c1ddf418593db88d084b684d1 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7a4bee67bc5d91bf7387c5af24b3cbe430d9ce80","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-09-09 12:56:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb850b130d05702e29afde50954cc0bc7d89bb85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 14:59:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd657fe530f4723bf6100b6ac910c52 : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ef6e91d84e14541b380a3ec3bb07fa2 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce691d0b659349b8b442871052d57bf7 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0014bc5ac59546c2a8545614aae18ff5 : SUCCESS in 10m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d59a5860fe448ef926fcc1b64c5f998 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00d71434f1b4eea881f8fcef8aeeb82 : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae262bdd1ee437b9d41b502da8bf7d5 : SUCCESS in 5m 18s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c4d6088cd5b34358a6b11156db33ecbd : FAILURE in 39m 11s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4aae5364a904407199808bdc01a83488 : FAILURE in 44m 44s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5881fb10d1b645cea98e4d13cf74603c : FAILURE in 48m 42s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/977e3e53bfb0467d8574126891a8cbda : FAILURE in 1h 48m 33s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/049034d8658d42199ee7aad6bc8c7ff6 : SUCCESS in 1h 45m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c162ed3cdd97485cbe2c268eedb7bc7d : FAILURE in 44m 10s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/eabd86880f5640b3a43878817d0a9066 : FAILURE in 41m 59s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/526d18fe1aa548d78e23ae24f9ad3192 : FAILURE in 43m 34s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/40c63cec68684bd0b77f25ca965620cf : FAILURE in 45m 50s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/27dccf0f994d43e08e13a8e3b8c91cdd : FAILURE in 42m 05s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8a10a7c2e2814c919d15578ea58280ee : FAILURE in 46m 07s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0785920bbb4449c58e3536bc364c3c9b : SUCCESS in 4m 24s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/3d9d0652e51b47cdaec35f9cc7c41033 : FAILURE in 4m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48beb78028f242a5a4a64769b5f706e8 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/503b49c7c406424c9d4399fb2d4689d4 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f16693c21b41f5c8f2e0195a117e9fe6559d5b76","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-09-17 13:54:04.000000000","message":"Patch Set 2:\n\nrecheck (no space left on device)","accounts_in_message":[],"_revision_number":2},{"id":"7d22cfed7cfe756afb410168846c757c2ec1e494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:59:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31b05ff28a4844269cf2dc6cabda21de : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8c72cbda3774d87a05ba3023372c12b : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc0be9bd52a49018d704378845f6067 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f46f2f21f3b417e867aee4f150f0f6e : SUCCESS in 10m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccefa58f661d458eb61a6f1d2f339859 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32e71fd058f8424b9802b379a4055ada : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030cec1fc6e94461afba2bc3de5d079a : SUCCESS in 3m 33s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ebb8fd8fa79441618b357ad5426fe2f9 : SUCCESS in 1h 58m 54s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/581d238d00694c2094d8fd88ddd2a500 : FAILURE in 1h 02m 37s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dbe2ff6440a243cca0aca5ee62e7e14d : FAILURE in 48m 23s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c82f91f524294948b3b2282659afa50d : SUCCESS in 1h 56m 58s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/44feb96dd8734b78b44e9e92a76a4125 : SUCCESS in 1h 40m 18s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4b9c1b58292140439938b01afd936dfb : FAILURE in 1h 09m 09s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/1c727e721e2e4049be9980b238baa71a : FAILURE in 47m 56s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/2056578420ab49abb7beccdad9acb64f : FAILURE in 55m 43s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/fc50adf2dd094ec1b32367643db2fc23 : FAILURE in 46m 23s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/361dc41d54844b6eb8e49f4f397889de : FAILURE in 50m 39s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/22b9e49748f241bb822777cea5f5c995 : FAILURE in 43m 04s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/bb59104c0f0b4d62929dadfb5eae32f3 : SUCCESS in 6m 25s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/af818d7328234c1d981e8c5843cd1e44 : FAILURE in 5m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22d85f3c59e342d2890a57f73c55a6b8 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14f8cce1df454efbbd551b4c393dde8c : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":2},{"id":"50e2694055eb6b5fa1f18b357bba8a0a88862ddf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-02 21:35:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"765eb50a981be4487a8425bd9019f46b6562c6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 00:42:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/787f7eda2879495dbbb7d825d9f2afc0 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f59c4a37a9bc4bf889c68f5d58b5d5c0 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d467ceac1ff4d4fab3d0de7e0ea2f1a : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb8e8793f040489285ac9c18316fa6ff : SUCCESS in 10m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9bd4dbde57164d24b92873d1822117fc : SUCCESS in 24m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936df692f0de4f38898600a75c2cfc6a : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fadfffa9489b48808a94eaae7aa98a3b : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc8f8686a01841fe90e3725c8715f5f8 : SUCCESS in 11m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/38d9a6aaf1b54f56b412ba557eb70c61 : TIMED_OUT in 3h 04m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/cb09b71454a34b68b5bb42150ae4cdc3 : FAILURE in 1h 13m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/2ad6dc7183c9427e963402f386917d37 : FAILURE in 1h 09m 57s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/491aee801c7a46c69102c02a05306718 : FAILURE in 1h 55m 17s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/879b7cfafd2c4d4db68e3d3d6eea4378 : SUCCESS in 1h 56m 38s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4b7bce4c49c540f1832d8f81b2150d8e : FAILURE in 53m 44s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ebcc553e52644e079ad1d4211da26b17 : FAILURE in 48m 50s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/80c0eaf6b1b343798a4b331600a01799 : FAILURE in 41m 11s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/95ad0bd8fa1c48bc9394eefc05cdf8e5 : FAILURE in 42m 21s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/5b77123aaa8f4d2da3a154e44aa8210c : FAILURE in 45m 29s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/577ea341a96c4aa3ad8c8597ab12eb72 : FAILURE in 43m 19s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/42d984a9363845a4a802b364af435d59 : SUCCESS in 5m 14s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/1ad3085e9e194b03a1d652d209c05442 : SUCCESS in 4m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5c1189047374f55917065146e199be8 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fd540596f71488bb886dd06a757b681 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":3},{"id":"bd1ab9838b3cc9a73333999e7c705a4ee62ad1b3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-03 06:48:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e2db62092fd56491f128f20b7cad7cb46b2ea3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:35:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/303db0a5d79d45f3b0d3b0f79aa80dce : SUCCESS in 16m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e9e537562e6494ebb85652f41f6645d : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2f826707cb48ababfc34186c24bf95 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81d8baf37c1d46589cc1d3872b2cbac7 : SUCCESS in 9m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d0f7178ec36843429fc5048a73cc8b48 : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbbff140dff64d65b5c32ca9f1bf6d34 : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f207ef323feb410c868db37da8644f5c : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab648233acdd4f879601e17ada261ca8 : SUCCESS in 10m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ca8e4bba2c094895ab17143c3a674787 : SUCCESS in 2h 45m 00s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/65f52239f74f400baa966eed4e281aaa : FAILURE in 1h 00m 38s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c842ec2dfcb646c9afba82c9de7c5740 : FAILURE in 1h 00m 37s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d2a7febb215843caa3d1a405fba0d998 : FAILURE in 2h 07m 52s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a3f4ff3fe47c45c9b38978606e220c18 : SUCCESS in 2h 04m 17s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/f0d3344dff80417d916810be8472f40c : FAILURE in 58m 58s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c5b5d27da5f247ca98bdcdf2ee15a96c : FAILURE in 39m 34s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/33934974e8044a9faf7f80c1d1c5db7b : FAILURE in 58m 42s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/7b4695425c9d4e2391acd107a260b702 : FAILURE in 44m 06s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0371e34c33a4436798947e84e82cf8ba : FAILURE in 41m 16s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/ae79c7b76ce941c995e41e84cd1d786a : FAILURE in 53m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d4fda2e9eb904eeca2cb3e99b80aa1bb : SUCCESS in 6m 44s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/0f1ca53907234e61be9e20ac1c3b676c : SUCCESS in 4m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17c867492a1f4232af94119bb8d1e993 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/44a1f25d0d82492fb1307da8893449d7 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":4},{"id":"efeaf498c126227ae14774999bb9a50da576362d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-03 10:19:50.000000000","message":"Patch Set 4:\n\nThis patch needs to hold until https://review.opendev.org/#/c/686373/ is merged and a new python-troveclient is released.","accounts_in_message":[],"_revision_number":4},{"id":"595affd8f5b08181fe495db3c9c1310434729df5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-03 11:58:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0f638a94681de26e5a8a278639b3679dcbfff478","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-04 01:22:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"438f2c6faea08248b0ec64c0119e056b935220f3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-05 00:15:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b044efa44f3237a388b1d6dad26a8c7cb175c84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 02:38:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a015f1779cb949cf873898e8cd84c424 : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55000c1b26e45598a517d3dc176b7b7 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4694a6b5a5842f292dcab371c6e9bfd : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b611d123d57246329562a3d09d084e28 : SUCCESS in 10m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2fd82058ecff4b1fb43d239ae5bc7038 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1aa66c2ba0841dbae29fd9a047284f8 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f3eea594b14a11a252e06c0eaeb37b : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c360339e4fc94ee0b720b22f295f470b : SUCCESS in 6m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/eebc6b5ce9c7432496c5d7b3087ae6a1 : SUCCESS in 1h 47m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bd69825991f54307a99116ed641f50e9 : FAILURE in 43m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dd7c047381b946e6a839c525b715c010 : FAILURE in 50m 55s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6a540a6f0d214f0c8ce86d8f063b0053 : SUCCESS in 1h 43m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c73ad9b0ea634acab8ed96c75fb81f4a : SUCCESS in 1h 27m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3370d11c1f764527af0bf045296f266c : FAILURE in 40m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f1bc6b25d4a84d73893c31fc5fa17583 : FAILURE in 43m 38s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/522cf309e3834274bb2dd3be12b69a04 : FAILURE in 35m 23s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/27737acf80394058996548966982d211 : FAILURE in 46m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/541a4eeb2d1e4d4e8f4bfe3dc59f28f5 : FAILURE in 39m 47s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/f2be097511dd4dcb9e6922c7fe300e7d : FAILURE in 50m 40s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/15f1af8b1ca8416c889c874552eff2ed : SUCCESS in 5m 36s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/213701fbb3f54fd18db454daddddd9e8 : SUCCESS in 4m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a92906ef48f412fa580e8015520514d : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95a15c3684564706b3b1c6bec26533ab : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":5},{"id":"dd6644c6816756da4974d8c55ff551ab0e7ce4d5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-05 08:32:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dcbeb230198b92d9db42b8c704383b9198de1275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 08:33:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a23af24be25e6c203e176f9753c01fdb6c405100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 10:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1b74c685421ac082d0ec10eaadbd77e0118ff4cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 10:19:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c29f23aa6a8a453a88657852170a4fd1 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951badca6e2442fda5fbf1b8207a72ce : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50fc60c86155406f9860bacc8e176c6c : SUCCESS in 9m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9cc7a85719b14421acc0befd8a80348c : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e12abc49fe4f424984236b984b99a67c : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c396a657abb34ddbaec2e020eeb08c2a : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/373a4c2bab6f41c399f1677ad1975b58 : SUCCESS in 5m 59s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/a48f0ca42974401680db93d13ad51f2f : SUCCESS in 1h 41m 24s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/52b5a282838e46daa0967f1c67bce68d : SUCCESS in 1h 25m 47s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/103e0b9276734b1c8881d7b30756356c : SUCCESS in 1h 26m 21s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/cfc1d6f75eb746d4b8ed50d22a4fa7cd : SUCCESS in 4m 01s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/8441a31dc570432fbc305bb22b044088 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9eec871629c140ba85d0f33ba192f20c : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":5},{"id":"3188668dc64d03a1bbdaaa4aca93aef1d1a7c3d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 10:20:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7a4e96acd04f28ad03cc9af6cb044e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5acf8bb1f5a245f080232b4bf7db55e8 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"67566d07b5308ab7ad1cc91729162b65d8da8638","revisions":{"0415c9e13a8b4d9ff16bded133343b5fb6d4fbd9":{"kind":"REWORK","_number":1,"created":"2019-08-21 15:42:39.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/677768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/677768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/677768/1"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-21 15:42:14.000000000","tz":120},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nConflicts:\n\tdevstack/plugin.sh\n\tintegration/scripts/functions_qemu\n\tintegration/scripts/trovestack\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n(cherry picked from commit 7b3483723a5fdfa56a8cc419ad77c8660ac87ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0415c9e13a8b4d9ff16bded133343b5fb6d4fbd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0415c9e13a8b4d9ff16bded133343b5fb6d4fbd9"}]},"branch":"refs/heads/stable/stein"},"66d0a9d442b0f29b9bd291a0cf08211228595694":{"kind":"REWORK","_number":2,"created":"2019-09-09 12:56:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/677768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/677768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/677768/2"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-09-09 12:42:20.000000000","tz":120},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nThis is a combination of 2 commits, both are necessary to fix every CI\nissue:\n\n1. Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nConflicts:\n\tdevstack/plugin.sh\n\tintegration/scripts/functions_qemu\n\tintegration/scripts/trovestack\nNote(elod.illes): bullet 3. skipped from backport as it is not a necessary\npatch I think.\n\n2. Use opendev.org instead of git.openstack.org\n\nAfter moving to opendev.org, the following command fails when building\nthe Trove guest image:\n\ncurl -o /opt/stack/devstack/upper-constraints.txt \\\n  \u0027https://git.openstack.org/cgit/openstack/requirements/plain/upper-constraints.txt?h\u003dmaster\u0027\n\nhttps://opendev.org/openstack/requirements/raw/branch/master/upper-constraints.txt\nshould be used instead.\n\nOther places using git.openstack.org are also changed.\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n(cherry picked from commit eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c)\n(cherry picked from commit 7b3483723a5fdfa56a8cc419ad77c8660ac87ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/66d0a9d442b0f29b9bd291a0cf08211228595694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/66d0a9d442b0f29b9bd291a0cf08211228595694"}]},"branch":"refs/heads/stable/stein"},"fc4b7286f9c6e45dd038dabee408d41f7dc35629":{"kind":"REWORK","_number":3,"created":"2019-10-02 21:35:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/677768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/677768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/677768/3"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 21:34:40.000000000","tz":780},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nThis is a combination of 2 commits, both are necessary to fix every CI\nissue:\n\n1. Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n6) Don\u0027t use master branch python-troveclient for stable branch Trove.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nConflicts:\n\tdevstack/plugin.sh\n\tintegration/scripts/functions_qemu\n\tintegration/scripts/trovestack\nNote(elod.illes): bullet 3. skipped from backport as it is not a necessary\npatch I think.\n\n2. Use opendev.org instead of git.openstack.org\n\nAfter moving to opendev.org, the following command fails when building\nthe Trove guest image:\n\ncurl -o /opt/stack/devstack/upper-constraints.txt \\\n  \u0027https://git.openstack.org/cgit/openstack/requirements/plain/upper-constraints.txt?h\u003dmaster\u0027\n\nhttps://opendev.org/openstack/requirements/raw/branch/master/upper-constraints.txt\nshould be used instead.\n\nOther places using git.openstack.org are also changed.\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n(cherry picked from commit eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c)\n(cherry picked from commit 7b3483723a5fdfa56a8cc419ad77c8660ac87ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc4b7286f9c6e45dd038dabee408d41f7dc35629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc4b7286f9c6e45dd038dabee408d41f7dc35629"}]},"branch":"refs/heads/stable/stein"},"799070da9de2f3a0dd959981fbe461d83cf58f23":{"kind":"REWORK","_number":4,"created":"2019-10-03 06:48:24.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/677768/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/677768/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/677768/4"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 06:48:05.000000000","tz":780},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nThis is a combination of 2 commits, both are necessary to fix every CI\nissue:\n\n1. Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n6) Don\u0027t use master branch python-troveclient for stable branch Trove.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nConflicts:\n\tdevstack/plugin.sh\n\tintegration/scripts/functions_qemu\n\tintegration/scripts/trovestack\nNote(elod.illes): bullet 3. skipped from backport as it is not a necessary\npatch I think.\n\n2. Use opendev.org instead of git.openstack.org\n\nAfter moving to opendev.org, the following command fails when building\nthe Trove guest image:\n\ncurl -o /opt/stack/devstack/upper-constraints.txt \\\n  \u0027https://git.openstack.org/cgit/openstack/requirements/plain/upper-constraints.txt?h\u003dmaster\u0027\n\nhttps://opendev.org/openstack/requirements/raw/branch/master/upper-constraints.txt\nshould be used instead.\n\nOther places using git.openstack.org are also changed.\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n(cherry picked from commit eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c)\n(cherry picked from commit 7b3483723a5fdfa56a8cc419ad77c8660ac87ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/799070da9de2f3a0dd959981fbe461d83cf58f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/799070da9de2f3a0dd959981fbe461d83cf58f23"}]},"branch":"refs/heads/stable/stein"},"67566d07b5308ab7ad1cc91729162b65d8da8638":{"kind":"REWORK","_number":5,"created":"2019-10-05 00:15:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/677768/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/677768/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/677768/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/677768/5"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-05 00:14:59.000000000","tz":780},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nThis is a combination of 2 commits, both are necessary to fix every CI\nissue:\n\n1. Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n6) Don\u0027t use master branch python-troveclient for stable branch Trove.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nConflicts:\n\tdevstack/plugin.sh\n\tintegration/scripts/functions_qemu\n\tintegration/scripts/trovestack\nNote(elod.illes): bullet 3. skipped from backport as it is not a necessary\npatch I think.\n\n2. Use opendev.org instead of git.openstack.org\n\nAfter moving to opendev.org, the following command fails when building\nthe Trove guest image:\n\ncurl -o /opt/stack/devstack/upper-constraints.txt \\\n  \u0027https://git.openstack.org/cgit/openstack/requirements/plain/upper-constraints.txt?h\u003dmaster\u0027\n\nhttps://opendev.org/openstack/requirements/raw/branch/master/upper-constraints.txt\nshould be used instead.\n\nOther places using git.openstack.org are also changed.\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n(cherry picked from commit eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c)\n(cherry picked from commit 7b3483723a5fdfa56a8cc419ad77c8660ac87ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67566d07b5308ab7ad1cc91729162b65d8da8638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67566d07b5308ab7ad1cc91729162b65d8da8638"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
