)]}'
{"id":"openstack%2Ftrove~678459","triplet_id":"openstack%2Ftrove~master~Ie7a8b4ccc90dc83fa7ced1b3c6361b9ff81f8d83","project":"openstack/trove","branch":"master","topic":"release-note-backup-filter","hashtags":[],"change_id":"Ie7a8b4ccc90dc83fa7ced1b3c6361b9ff81f8d83","subject":"Release note for backup filtering","status":"MERGED","created":"2019-08-26 03:53:02.000000000","updated":"2019-08-26 04:35:03.000000000","submitted":"2019-08-26 04:33:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678459-1566794021084-30e5c917","meta_rev_id":"ed8ec28ae0eeff06df584b83136a4a3f456137a9","_number":678459,"virtual_id_number":678459,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-26 04:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-08-26 04:24:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-08-26 04:24:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 04:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"938228d8c3b5f78af893116bc078586a75536c34","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 03:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690b1e62e4327c6579d5e45d65474f653d7a3eb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b465642f4a41e3b2efa0016302a4c6 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e74d6b45764981ab09590d1f6040c7 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009643a42001477385dfb4e9d8da71dc : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ad8eb0033be0a9aa3ee08d2fb30a23a51b63c838","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 04:24:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43be477fd667d962c094c3527f3109d65b01473e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:24:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c20dbbe5c2681ae9bcaeeafcc7c44b0ca7ddfd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:33:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2178642858e04e92bf933239c7b3c77a : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982abe2e9152414ca948be535ff953e4 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/884e830b510c4b708c749807d394b642 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d839c26a731a0a2e9ba34afe9b846082f703d518","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:33:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5adb6099376d926e783fd10e3ef937fc7d78f00c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:35:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a30d5fbb5da42a5bf18d3f429d284b0 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"858be706578af581526fec7ea0d4024bc930f870","revisions":{"858be706578af581526fec7ea0d4024bc930f870":{"kind":"REWORK","_number":1,"created":"2019-08-26 03:53:02.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/59/678459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/678459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/678459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/678459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/678459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/678459/1"}}},"commit":{"parents":[{"commit":"77dab835e6a1bcfc9ca104d72015c69cc17318f6","subject":"Merge \"Improve devmode\u003dflase when building the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77dab835e6a1bcfc9ca104d72015c69cc17318f6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 03:52:31.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 03:52:31.000000000","tz":720},"subject":"Release note for backup filtering","message":"Release note for backup filtering\n\nChange-Id: Ie7a8b4ccc90dc83fa7ced1b3c6361b9ff81f8d83\nStory: #2006433\nTask: #36345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/858be706578af581526fec7ea0d4024bc930f870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/858be706578af581526fec7ea0d4024bc930f870"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
