)]}'
{"id":"openstack%2Ftrove~678461","triplet_id":"openstack%2Ftrove~master~I62323ac1c1edfbf0378d0c89e156fdb3a879aa88","project":"openstack/trove","branch":"master","topic":"guest-py3","hashtags":[],"change_id":"I62323ac1c1edfbf0378d0c89e156fdb3a879aa88","subject":"Support python3 in guest agent","status":"MERGED","created":"2019-08-26 05:53:26.000000000","updated":"2019-08-28 01:56:06.000000000","submitted":"2019-08-28 01:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678461-1566957280859-6ccbb2a0","meta_rev_id":"1f01356e50ebfe749e8e7a8db37484608b141169","_number":678461,"virtual_id_number":678461,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-28 01:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-08-27 11:22:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-08-27 11:22:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-28 01:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62729199b28e920bfa7cd50e49fb6d54054a4f19","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 05:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05334696871e370bdae1ea723eb2a1e8b3434f2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 07:15:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0d9197e6ad6476f9fe1866ce38ceb4d : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8df2cc677d4e4cb1bfe424f903c5a38b : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75d3c45de6f4b2fb7f9abe94a5cc933 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f2f4018b02c4e198bcae237ccfa27db : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d78e11259f400a9140176104cbe0f7 : SUCCESS in 14m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbeb77d1aaba4fe7b270c5159bf527c2 : SUCCESS in 16m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902d2e51bc9843bba9cfe51595793c75 : SUCCESS in 5m 57s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8bffa1a1508c4fd6bc674ac228dc1ac5 : FAILURE in 56m 11s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/426d0b1afb78479f84415dca1780c58d : FAILURE in 39m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/04d7a57b74be4a4aab401054b1a6fdab : FAILURE in 40m 53s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/57864a8dc29944acbb8ca1554ad92e2f : FAILURE in 1h 14m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fa4b2e72f3304b1bbc9e39eb28fe0e0b : FAILURE in 55m 59s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d17028eb454d4cdfaa8427e2138a7548 : FAILURE in 36m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/947e65fbc55c4ec2a7b1f6fc9b14acf4 : FAILURE in 36m 56s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/1ab28376ad0f4591aeeab7d6d204ffa1 : FAILURE in 39m 30s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/2aae96fdc6f24ca2bd4b3ac2cd519a76 : FAILURE in 43m 01s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f7e01a08fb134f4da39f2c23f4590e25 : FAILURE in 40m 51s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/846cf68b5b884043b2e93dbb5f6e309e : FAILURE in 42m 19s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/8b6f54976a1042b29228ca9535e50e81 : SUCCESS in 6m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96215db613454fd9bd4880c1e2a4f837 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e05ff24832544a5793f1585987c73b9c : SUCCESS in 8m 27s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/335ed4f5c6414ebb8d4f34c01bf1378c : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c7646f1ed67c5d3b90ad62b9eb5739534cd1722f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 08:02:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a563c4d7365c13bb7863e2552e1087ea6bfe192d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 09:25:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf5d7f102c1544efa73de01e320659b6 : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55e0d9fa79834b5f903d73c4f4199197 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a40654a9114e5c994989b78156722c : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e32cf2e9f7c4e5f8936d7add8ed2bcc : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a2479d4d25c4df68f6c0d116ff8d90f : SUCCESS in 13m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7058316992f840eeb6eb0fd15d615f58 : SUCCESS in 18m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d563e80529604151a5cad91c54b3cc74 : SUCCESS in 6m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/43a6edd6f2664117bda83a818592545d : FAILURE in 1h 06m 31s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3b1398bdadca42c498353475979ee43f : FAILURE in 52m 22s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dc0965c82d35425588c1c0b6a1d2bb14 : FAILURE in 45m 35s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5c1fb337c11c4aef9bc6b298d3a1a81e : FAILURE in 1h 14m 47s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e5266f650d8d49dc9cb185109ae33a79 : FAILURE in 56m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ede6525c47a841769e4b394d87b877d2 : FAILURE in 46m 16s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7d7b7dd318ff48e8b3058b3a1517a517 : FAILURE in 48m 43s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/58e2195036114bdeb4d575fd334d6117 : FAILURE in 41m 53s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/b20139285a5f4d9d8702aa20ac0501de : FAILURE in 44m 17s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/b2cc18d111f2426b85043f5931958d81 : FAILURE in 43m 06s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8424a11820e94ffda9c55b1d72bdfdba : FAILURE in 47m 23s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2d88b21a22394a858931ed46b1de12c6 : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e743c8a9d8af444f8d6d446e81b8d3a3 : SUCCESS in 16m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ccf76438721d4b259e6314a05fb2e733 : SUCCESS in 8m 51s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fc111574f7f4e809253a719dec0c7fe : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":2},{"id":"efbac82fd9b9cfbe315876932b1a665cd12b29fe","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-27 04:06:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81d2c2c41c51e96c519a7679cbf6f637b360bd8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 05:59:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8476c065d7d4d3bb798b1fa944ba02e : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54ccca8bb79d41058e4b04ef4901b5f5 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807aecd239184fddae67cc4db43b1490 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de79ea39e8dc4969acd2eec38cced3e4 : SUCCESS in 9m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d1e0b75a28341f09818a1ecf93d38df : SUCCESS in 13m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25936fd34a70417494def5f9b8a81430 : SUCCESS in 18m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40cb2011494418696e717c6e16fa7c3 : SUCCESS in 6m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6d95fe54e832457f88ee5496d341e925 : FAILURE in 52m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/423a2963b5884f3fb4255f0a3e14ef0c : FAILURE in 40m 57s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/102931ca7fc442fd83d45a676eb98056 : FAILURE in 49m 30s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ea5fc03f75904b88b4e5847ead717ef7 : SUCCESS in 1h 46m 49s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9f2a15f8b56f489f9028bf734f298201 : FAILURE in 46m 53s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ec773b494f1e459ca5a01c2a66baf989 : FAILURE in 39m 46s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/db91823ba8b340858752f0243cbee988 : FAILURE in 46m 29s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/a5fca946f8234718b244c7acfebdad82 : FAILURE in 41m 42s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/e3d1c3ce25274ed580091294d4e498cb : FAILURE in 41m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/2ad5c6525efe480299506399e4d0f005 : FAILURE in 46m 44s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6da30ed4d93b455b91c44b1f670a48f2 : FAILURE in 40m 40s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/45b4951d3ec64f2c816039d55cbdba22 : SUCCESS in 6m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1288c846d552400c812dff8c90ce56fc : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9844f2d6eb5647c6bb72c225a9a95494 : SUCCESS in 8m 50s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4149a2b33f01413694d375df4222ba5d : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7c640ac061dc76ad97271bc0e895a6459e8665e8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-27 08:40:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6a8f33ff39b09660979c56f54ae884280da5dc7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 11:13:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6b7c407c7d24a6cb81a4b2eba317d83 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91ad63b1322748998f5e9a04f2022e0e : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a1e0d3d6ab4ff8b5a5a41f0d7b9e50 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00b3bc4f28474758a5e8329fd619e491 : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/337afd7d7d454516974e571d65366d76 : SUCCESS in 12m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceeba0530dcb42869e8b328cd50a5ce0 : SUCCESS in 18m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2024eca6fe14480a90df1ca40447eac0 : SUCCESS in 5m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/058515d1a21e4e809444c922796c6830 : SUCCESS in 2h 26m 57s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/86d39e745d4b43caa650c6bc84567c79 : FAILURE in 44m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/da2a315428a7478c94c6d0f082274ca7 : FAILURE in 41m 18s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/88a666453f8f45538f482c61957dbaaf : SUCCESS in 1h 44m 14s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/51a4aabdce354bc6bbf7022ce17990db : SUCCESS in 1h 53m 20s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/0684c74080e14fa980ac4b0b7d6133f6 : FAILURE in 41m 28s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/52b92803093b4f628bfa171aac3431a6 : FAILURE in 45m 33s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/1444a96aefc5491bb054c6e845074041 : FAILURE in 47m 28s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/080ae3174cf64372bfcba801271401b9 : FAILURE in 40m 43s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/441b72f781344c26a7ee8769227f2267 : FAILURE in 44m 42s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/5bfd9d42282c4fb9aeeb5f2395209cfa : FAILURE in 48m 18s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/634bc746c0894d619810105bb69f6345 : SUCCESS in 5m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/007f27b4c2c74fd9954eb243567b6c51 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/952bd58299c04033b467654fe1158a13 : SUCCESS in 10m 04s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1f25e82bcc640499465f4364260e9ab : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":3},{"id":"bb477a34a14c781ea81bb66e83ba12d42a1b2fd8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-27 11:22:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"64a80619d624c74c6996eef68abf3473d85c353c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 11:22:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bdec8e0e45140fbc4df7ae46f569703fc6aa2534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:45:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6654b4450655476da0bbc09cbf1d492e : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61333d0475544c3392c00fa7e36184df : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10edc13b6de44c3da9a2635b49b76ed2 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac36bd07fc240e58f9dd1cea35d8b5b : SUCCESS in 14m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a59227e764b043edbdbdba8fafddd77d : SUCCESS in 20m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c364d08f4f4da4b8b357f3a96b682f : SUCCESS in 8m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/74494adf9d434d6392266ab9ce2dd736 : FAILURE in 2h 21m 01s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/81be414a33f44f848cfc3f6039ad5a9e : SUCCESS in 1h 46m 09s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f740f4fa0e614fb2aec43c44c06ceb77 : SUCCESS in 1h 48m 12s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/3b845c8e2bf54483838f9347285e9e77 : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4727ef2fe97477597b0a99ec1b2f023 : SUCCESS in 11m 19s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07a159b2273549479ee98fca7729d532 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bae30109b2de8c6558e6f4f5561358ab7d390f91","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-27 20:55:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6c78a7816f2a9e6d86844076d3d6aeb990e6aff5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:25:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b3c895c92140eb96d9670e8baf7181 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f4490f929e43ffbe1c2abf01d908f8 : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ee2cc972264d1f85ccf51f7ad50025 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e5d93a5a6f04a53bd6d3b726076ca93 : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7faac437f03435ab20149a5665a12b2 : SUCCESS in 13m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6574ba3f9384a638b5e30d179a795b5 : SUCCESS in 21m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3208e94de82b4d9aa65db715f9ac3e71 : SUCCESS in 5m 41s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/aaae64ac11d34d44941611e93ae40183 : SUCCESS in 2h 23m 00s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c05a080ea6c548d1941af56322d896c1 : FAILURE in 49m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c91ba4499d4e4ac58d8408583baffedc : FAILURE in 44m 06s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/412e26d613554a35afd565ece0884e73 : SUCCESS in 2h 01m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5ed8ae6d1612441ab17e2293eb23a14a : SUCCESS in 1h 52m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/97a763ac975b4491ad38072dbb955364 : FAILURE in 45m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f9d550a5e9eb487883e9ca9c6047d048 : FAILURE in 46m 54s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/077ac4e902aa47cf9543159a853586b2 : FAILURE in 48m 00s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/1af9797d3e914346a5e8503209976bd2 : FAILURE in 48m 46s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f7bb737721e6427386b62b6b0cea96f4 : FAILURE in 48m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6116e3dbafa243389bf75b86d529d8c4 : FAILURE in 42m 43s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/af486d736589433da20f2bd7ce8dd371 : SUCCESS in 7m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f8b862b5f064621a54ef98b8afd9798 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7e702e1524846029ae55ff42c287ca3 : SUCCESS in 8m 27s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c281f4c50484465ba79a35502981154 : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cd99519195e8ed557d3b7b5eeaabb157fbfc3d1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:25:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9baf507d2cf576b3f066752cf7a3546023a28819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:54:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d5d47b1ba064bd8ae0eaa1345d8450f : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b23bc5192a4460ba1fac286e0f81865 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9529e759e9540cf8fac08aaad9118e2 : SUCCESS in 11m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/611407fbe35a4966b0853734501bf69f : SUCCESS in 13m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee911034fae94ba183dfac7088f7b3ec : SUCCESS in 18m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c0d0784ecf4bcea9b0e0789da57985 : SUCCESS in 5m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b872df875a624169ac47094fc598a0bb : SUCCESS in 2h 23m 57s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8b0c9648804d405cbbba0dc10c8bf7e5 : SUCCESS in 1h 50m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e0bbfbe590474252ab50dca20e2b1428 : SUCCESS in 1h 48m 34s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/4ab8c7eb271e42e4aade36ecb1e0d844 : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8746d97fdabb47a3b60f55573812ccbc : SUCCESS in 8m 15s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdff403eff2e4707b423c579161f5065 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":3},{"id":"81c284c7e1f4695a4a50701f0e8fe080a7d3a8ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:54:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"85a48625296e8cf17d986157685cf4ef5dee67e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:56:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763b49a4a2c14a98aa6bf83c5d6b1668 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7b9a752eb09083f8171c6ccfa3870bdc8032613d","revisions":{"1fd0c4978da005dfbecb7511256cd0e50947cc90":{"kind":"REWORK","_number":1,"created":"2019-08-26 05:53:26.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/678461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/678461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/678461/1"}}},"commit":{"parents":[{"commit":"77dab835e6a1bcfc9ca104d72015c69cc17318f6","subject":"Merge \"Improve devmode\u003dflase when building the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77dab835e6a1bcfc9ca104d72015c69cc17318f6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 00:10:32.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 05:52:53.000000000","tz":720},"subject":"Support python3 in guest agent","message":"Support python3 in guest agent\n\nStory: #2006455\nTask: #36376\nChange-Id: I62323ac1c1edfbf0378d0c89e156fdb3a879aa88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fd0c4978da005dfbecb7511256cd0e50947cc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fd0c4978da005dfbecb7511256cd0e50947cc90"}]},"branch":"refs/heads/master"},"8433e2539d559204649da6254fa1902bd817b30c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-26 08:02:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/678461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/678461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/678461/2"}}},"commit":{"parents":[{"commit":"77dab835e6a1bcfc9ca104d72015c69cc17318f6","subject":"Merge \"Improve devmode\u003dflase when building the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77dab835e6a1bcfc9ca104d72015c69cc17318f6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 00:10:32.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 04:20:08.000000000","tz":720},"subject":"Support python3 in guest agent","message":"Support python3 in guest agent\n\nChange-Id: I62323ac1c1edfbf0378d0c89e156fdb3a879aa88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8433e2539d559204649da6254fa1902bd817b30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8433e2539d559204649da6254fa1902bd817b30c"}]},"branch":"refs/heads/master"},"7b9a752eb09083f8171c6ccfa3870bdc8032613d":{"kind":"REWORK","_number":3,"created":"2019-08-27 04:06:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/678461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/678461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/678461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/678461/3"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 00:10:32.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-27 03:03:45.000000000","tz":720},"subject":"Support python3 in guest agent","message":"Support python3 in guest agent\n\nStory: #2006455\nTask: #36376\nChange-Id: I62323ac1c1edfbf0378d0c89e156fdb3a879aa88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b9a752eb09083f8171c6ccfa3870bdc8032613d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b9a752eb09083f8171c6ccfa3870bdc8032613d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
