)]}'
{"id":"openstack%2Ftrove~678516","triplet_id":"openstack%2Ftrove~master~I3487e45840f72875ffd6cfb6ca75b4bbf60ab51e","project":"openstack/trove","branch":"master","topic":"fix-volume-type","hashtags":[],"change_id":"I3487e45840f72875ffd6cfb6ca75b4bbf60ab51e","subject":"Make volume type optional","status":"MERGED","created":"2019-08-26 10:22:40.000000000","updated":"2019-08-27 04:35:03.000000000","submitted":"2019-08-27 04:33:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678516-1566880411320-ebf41b20","meta_rev_id":"04006ab3d823e6ed8ce4cc8b0b60a2567fdf427a","_number":678516,"virtual_id_number":678516,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-27 04:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-08-27 02:12:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-08-27 02:12:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 04:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cd0b68a8ca551aec11472778cf2fa737c223d56","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 10:22:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"004d20919fb547efc234330f38edba94aef44fc8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-26 10:24:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5dd46c5bffecf84eb9dd38dc46a24d923e5f8275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 12:55:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c322aa86b3846f7a9a7f54f93aeb881 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66b8824c644c4ba68d5f48d70ca0a32c : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ff7ea8f94d423b8d522e70d655c03e : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17bbd73bce16471c9e0304492258eb40 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/178e1ec74b9d4453852a9ecb4119d1d3 : SUCCESS in 11m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64e08599a72c4002a332f0fa7fa03b82 : SUCCESS in 17m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e6f97fa7d94fc3a7cf1668de7688cd : SUCCESS in 5m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5ec4a9a732e24a68a3e2ef3f4b7cc87c : SUCCESS in 2h 28m 50s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/824868fa3f80499496f5938bd90ad7cb : FAILURE in 39m 35s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3ba1b450aa9b45f285c06884df2d18d5 : FAILURE in 38m 07s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f2153ed4eb05452c8fd4e914e4a6dd9b : SUCCESS in 1h 30m 34s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0662d80c6abc4c39b99ca321313a90f3 : SUCCESS in 1h 39m 09s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/70c78a0fc03640c882561f975dbbcd21 : FAILURE in 41m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/eda27a785eae4b3f9625b203a34f7872 : FAILURE in 40m 12s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/c08298d215354a8c8e4643d8ccade1a6 : FAILURE in 40m 01s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/db3e4f4a0ee94b85983cf023f1f77bc1 : FAILURE in 41m 46s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9e84e0c82a34433faf1d209badcf7f4f : FAILURE in 1h 12m 02s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8ca8204f156e49a793347dc923c8ae3a : SUCCESS in 1h 24m 11s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/aa49352fdf8f4850b3352dc79ee237bf : SUCCESS in 11m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30a39beef7214bba8b9cc2076240bdcd : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d92a65b381b54dc89fe65466babe252c : SUCCESS in 8m 58s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f2138204683489bbd35a6dbd19e20b9 : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3a6a8141845ce49dad944cad29931a5dfc8f967e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-27 02:12:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f9b11bdd777636842839e9c7714d60364993200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:12:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5171c2b6247dbfbebea089527eaa2faa9436cde9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 04:33:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38ec18a2fa9649b0add7f7cced51960e : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2aedfea5fa14b0bbf8221a45d06dfaa : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/243f52729d6443298db647ac0bd85350 : SUCCESS in 11m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e206dfc8ed64490fb3ec5bea25082d35 : SUCCESS in 14m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b00d3d0afaf4c18a142956fd3463b10 : SUCCESS in 19m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd7d9f2065140a1a2e0b83dc52be0cf : SUCCESS in 5m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ef029483638b46d0b4f21e8c62afa343 : SUCCESS in 2h 15m 19s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8b22a495c76e4558bd70b6c2600c9f22 : SUCCESS in 1h 35m 06s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4b2edd49485a4de382b01a5ff8c606d6 : SUCCESS in 1h 48m 48s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/eda724e5c5f04fd3ad21dfa0e91763a7 : SUCCESS in 5m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2bc62c094a5481fa02151be6a6b546e : SUCCESS in 8m 55s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43fbab78cfd4ac2845f15467ed0bf18 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":2},{"id":"093079979d953ed232fca6bed297177b01f016e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 04:33:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"feca42cbcccfbc608b7f614a4d7d5e404c1468bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 04:35:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4cf65e254d4d4997ecb6e358a1cbdf : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ee891ddfc647907957ab51fbf225d9c0783eeaf","revisions":{"1578b965ad28f4347f9b4f9b02669156b0c01754":{"kind":"REWORK","_number":1,"created":"2019-08-26 10:22:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/678516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/678516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/678516/1"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-25 09:03:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 10:11:16.000000000","tz":720},"subject":"Make volume type optional","message":"Make volume type optional\n\nChange-Id: I3487e45840f72875ffd6cfb6ca75b4bbf60ab51e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1578b965ad28f4347f9b4f9b02669156b0c01754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1578b965ad28f4347f9b4f9b02669156b0c01754"}]},"branch":"refs/heads/master"},"2ee891ddfc647907957ab51fbf225d9c0783eeaf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-26 10:24:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/678516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/678516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/678516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/678516/2"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-25 09:03:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-26 10:22:48.000000000","tz":720},"subject":"Make volume type optional","message":"Make volume type optional\n\nWithout this patch, creating db instance will fail if no volume type is\nprovided:\n\n  Validation error: instance[\u0027volume\u0027][\u0027type\u0027] None is not of type \u0027string\u0027 (HTTP 400)\n\neven if `cinder_volume_type` is configured in trove config file.\n\nChange-Id: I3487e45840f72875ffd6cfb6ca75b4bbf60ab51e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2ee891ddfc647907957ab51fbf225d9c0783eeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2ee891ddfc647907957ab51fbf225d9c0783eeaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
