)]}'
{"id":"openstack%2Ftrove~678715","triplet_id":"openstack%2Ftrove~master~Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3","subject":"fix bug report url","status":"MERGED","created":"2019-08-27 02:44:44.000000000","updated":"2019-09-03 03:27:35.000000000","submitted":"2019-09-03 03:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678715-1567481147763-1d226e13","meta_rev_id":"58678287b0e3572856297f22a0c3e2799008bb80","_number":678715,"virtual_id_number":678715,"owner":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-03 03:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"date":"2019-09-03 03:04:27.000000000","_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-03 03:18:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-03 03:18:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 03:18:45.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-09-03 03:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b43b11fe42062403542661a99e1772fe487ad919","author":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"date":"2019-08-27 02:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3db24209f1fda1fedf51172f37111f9edd529c61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:51:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4033290253974affb67e693948404f25 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/357efc727efd4ce08e2b1471f2ebbfc5 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cc57f9c351415db480c947df54f98d : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"22b9433ec41fe8ae6b66d6b12e7474fbcddc0815","author":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"date":"2019-08-29 06:37:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"433b444d551850a87764a27bc817a958e227b9c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 06:56:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87d7e8c451b84310b4f031b2feb532cb : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c3bb77ba5e49989d62936475fb7640 : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a3c9af39744600acc7663e6da34459 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"91bb0e4188acfc7212c0d7bb77e7c5a16fd3c84d","author":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"date":"2019-08-29 07:09:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"cf04891e6434e650d6358926a88791c0bb8e3821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 07:18:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed80f47688924f9fad64ce7400d1b95e : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866494937d7c4662b854264968dc56d7 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3386e2cd417e4c26a3ad07ae971c8b61 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6656123919043e29f7d85bf25d1d337d0ea6ccf4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 02:44:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"55b85e5a1950476f36f13dc40567009adde9ba1f","author":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"date":"2019-09-03 03:04:10.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2398bc5120f0549ea30d2b51e0997c356b320a3c","author":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"date":"2019-09-03 03:04:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"970ef2acf1cc5303036522f6eb0858f48c7d3c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:14:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/016f1438169e4301ab5bd86dc3cda5cc : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9474015a1d4791b08ce4ad068b9cb7 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d776ab5b5a84440781378e6f9ba9b3f1 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"6cbe8726e4c2c86c2478fb5a1cfc56c66d192e92","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 03:18:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6f8d8ff14800121254d216ed672e7948160cc5d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:18:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87b03da8de48c07599124c11ef1a1b6ad8951772","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:25:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"80e832ac5c47ae8164715d54c40c7acbf4da3330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:25:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9efdefa73e45448c9690e0d736ad513f : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb0a726799a4aab8bfe11f769d7246e : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a0b03382e0416baca9ea2a5259968e : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":4},{"id":"58678287b0e3572856297f22a0c3e2799008bb80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:27:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71b1bccfe1c4d40ac2ede22b88cfe2d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1d3db54fbf73165d1fbe14ea9d715bd4d0cf3a1f","revisions":{"0227891d4213d3688aaa1b8488d04810936bc79a":{"kind":"REWORK","_number":1,"created":"2019-08-27 02:44:44.000000000","uploader":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"ref":"refs/changes/15/678715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/678715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/678715/1"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-27 02:29:58.000000000","tz":480},"committer":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-27 02:40:14.000000000","tz":480},"subject":"fix bug report url","message":"fix bug report url\n\nsince we don\u0027t use launchpad to report bugs anymore,replace it with\nstoryboard.openstack.org\n\nChange-Id: Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0227891d4213d3688aaa1b8488d04810936bc79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0227891d4213d3688aaa1b8488d04810936bc79a"}]},"branch":"refs/heads/master"},"64b01576e1b81a714f84417d7584cf0906c016c2":{"kind":"REWORK","_number":2,"created":"2019-08-29 06:37:28.000000000","uploader":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"ref":"refs/changes/15/678715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/678715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/678715/2"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-27 02:29:58.000000000","tz":480},"committer":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-29 06:35:12.000000000","tz":480},"subject":"fix bug report url","message":"fix bug report url\n\nsince we don\u0027t use launchpad to report bugs anymore,replace it with\nstoryboard.openstack.org\n\nChange-Id: Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64b01576e1b81a714f84417d7584cf0906c016c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64b01576e1b81a714f84417d7584cf0906c016c2"}]},"branch":"refs/heads/master"},"85282c58884492f2d3c982fadadf02b5e9d08d45":{"kind":"REWORK","_number":3,"created":"2019-08-29 07:09:40.000000000","uploader":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"ref":"refs/changes/15/678715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/678715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/678715/3"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-27 02:29:58.000000000","tz":480},"committer":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-29 07:09:27.000000000","tz":0},"subject":"fix bug report url","message":"fix bug report url\n\nsince we don\u0027t use launchpad to report bugs anymore,replace it with\nstoryboard.openstack.org\n\nChange-Id: Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85282c58884492f2d3c982fadadf02b5e9d08d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85282c58884492f2d3c982fadadf02b5e9d08d45"}]},"branch":"refs/heads/master"},"1d3db54fbf73165d1fbe14ea9d715bd4d0cf3a1f":{"kind":"REWORK","_number":4,"created":"2019-09-03 03:04:10.000000000","uploader":{"_account_id":30814,"name":"yangyong","email":"yangyonglc@inspur.com","username":"yangyong"},"ref":"refs/changes/15/678715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/678715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/678715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/678715/4"}}},"commit":{"parents":[{"commit":"40821578f9d3fd5264cfa1c1b1328425dea10f80","subject":"Merge \"Release note for backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40821578f9d3fd5264cfa1c1b1328425dea10f80"}]}],"author":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-08-27 02:29:58.000000000","tz":480},"committer":{"name":"yangyong","email":"yangyonglc@inspur.com","date":"2019-09-03 03:03:27.000000000","tz":0},"subject":"fix bug report url","message":"fix bug report url\n\nsince we don\u0027t use launchpad to report bugs anymore,replace it with\nstoryboard.openstack.org\n\nChange-Id: Ifa456c62c4fed197954cb7f7b2aaf0328a00d6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d3db54fbf73165d1fbe14ea9d715bd4d0cf3a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d3db54fbf73165d1fbe14ea9d715bd4d0cf3a1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
