)]}'
{"id":"openstack%2Ftrove~681372","triplet_id":"openstack%2Ftrove~master~Ie0c4ead89316001107c02450627eb5491ff8b521","project":"openstack/trove","branch":"master","topic":"public-db-4-release-note","hashtags":[],"change_id":"Ie0c4ead89316001107c02450627eb5491ff8b521","subject":"API doc: \u0027access\u0027 support for creating trove instance","status":"MERGED","created":"2019-09-11 04:00:01.000000000","updated":"2019-09-11 09:39:07.000000000","submitted":"2019-09-11 09:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681372-1568194747706-4c095369","meta_rev_id":"f08063f87baff82a272801f9003d40f6128d75f3","_number":681372,"virtual_id_number":681372,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-11 09:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-11 09:11:25.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-11 09:11:25.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 09:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"019122aad8efe5e1453d45b1c683549147c239c5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 04:00:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"decfa5e414cd6bfc6f1d76944b1864eb61a8ffe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 04:15:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c74ca4105b04c09a99ee415145a54ff : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580acd3455764eae953e46287492feb9 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f700b59ec997477da653954c60984456 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7c8846ed38ec97fbcfcc08797e5c594a93cc7dfb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 04:39:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca5ea88d5769982aa79f42811d58610ba584b4d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 04:50:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e4a97ef86bb149318bb49c0f90be75e1 : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2af60d59434876a17f497d51e345a5 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f3887348a74877864f40807c388961 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2098cbd6fd93f007316ed117308a90116f6cbb56","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 06:07:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5b1e6ab4d3b3111765a2f90405e453e2349ab32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:41:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9e4ce546ac74c7dbeca7ac9f3a20d0f : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79fc604c0c64b6b83b68e7fa9712314 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c209f18bef184e9b99566386fbc9884f : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"e071fbbf185e3120b068afa79420e760572218a6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 06:46:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ecd1bda6bc77f4912d14f96bb9c7223469047e59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:56:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6139369d3ce943dea73b4f3c0bf3a1f8 : FAILURE in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad01f2dc5eee4e7bb88d98523935ddf3 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e539162cb40741bbb6c65fcf380f58fa : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"334d2d2cfcdd8a34198565e0d7b7890d690b6604","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 07:09:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09a4c5cb12f0b25401bd377713a19ad467c223fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 07:27:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e8af7211302c4c54a5ae20406606abdc : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09146ba462404a4e96dca033c3184dbe : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31b62fe014542499ba283e14f320b25 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":5},{"id":"ef93cd344cd93773af668c93401fae0c1df897ea","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 08:45:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"317260bd5d0c9784d60e0a51e8f2a2e1f67ab23c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 08:57:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6543c525f1649baa7961ba853e2aff9 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3024b9fbba4e9b89e24901e7cb5419 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2689c5a41920437b8d82f4acb736ff3e : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":6},{"id":"0c6bbaa11b28597d9d17c90175ec872da178e764","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 09:11:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"992a8a9e5932715f7c0c632dd8481977bdf78a48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:11:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a0ff91de95aec8498bab9357838680c4ce2d4114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:39:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f87b0c5fd9282c2483ff497b678cf1d9651856db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:39:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7aea35305e7c43b3aeb090bf81ff3d94 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee0c3c627d5412cb6f65b547ba394f2 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2987e23bb3e34645a3a92e4b61d4b800 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"94dd0bd2b7cc57d69a3ba23345445d8b115600db","revisions":{"58bd857ca55698b77fb38d1e91a2b9a45832f5fc":{"kind":"REWORK","_number":1,"created":"2019-09-11 04:00:01.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/1"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 03:57:47.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58bd857ca55698b77fb38d1e91a2b9a45832f5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58bd857ca55698b77fb38d1e91a2b9a45832f5fc"}]},"branch":"refs/heads/master"},"1c890ebb25b9861d29d0f0e75b4e21b8cfa0ecb3":{"kind":"REWORK","_number":2,"created":"2019-09-11 04:39:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/2"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 04:39:04.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c890ebb25b9861d29d0f0e75b4e21b8cfa0ecb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c890ebb25b9861d29d0f0e75b4e21b8cfa0ecb3"}]},"branch":"refs/heads/master"},"991484ccac2ae4695b71613c8f9b4bc9bd33de1d":{"kind":"REWORK","_number":3,"created":"2019-09-11 06:07:10.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/3"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 05:50:35.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/991484ccac2ae4695b71613c8f9b4bc9bd33de1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/991484ccac2ae4695b71613c8f9b4bc9bd33de1d"}]},"branch":"refs/heads/master"},"39c58cb49006dc131ee210f299075e0568836eac":{"kind":"REWORK","_number":4,"created":"2019-09-11 06:46:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/4"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 06:46:04.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39c58cb49006dc131ee210f299075e0568836eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39c58cb49006dc131ee210f299075e0568836eac"}]},"branch":"refs/heads/master"},"b04228d4e8e3d4bb7f4b031054009badd754bdfa":{"kind":"REWORK","_number":5,"created":"2019-09-11 07:09:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/5"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:09:03.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b04228d4e8e3d4bb7f4b031054009badd754bdfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b04228d4e8e3d4bb7f4b031054009badd754bdfa"}]},"branch":"refs/heads/master"},"94dd0bd2b7cc57d69a3ba23345445d8b115600db":{"kind":"REWORK","_number":6,"created":"2019-09-11 08:45:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/681372/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/681372/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/681372/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/681372/6"}}},"commit":{"parents":[{"commit":"c33fa6706678131e7625a605e2f1dd3a2b0cd362","subject":"Support to create public trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c33fa6706678131e7625a605e2f1dd3a2b0cd362"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-08 23:26:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 08:45:20.000000000","tz":720},"subject":"API doc: \u0027access\u0027 support for creating trove instance","message":"API doc: \u0027access\u0027 support for creating trove instance\n\nStory: 2006500\nTask: 36471\nChange-Id: Ie0c4ead89316001107c02450627eb5491ff8b521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94dd0bd2b7cc57d69a3ba23345445d8b115600db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94dd0bd2b7cc57d69a3ba23345445d8b115600db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
