)]}'
{"id":"openstack%2Ftrove~681393","triplet_id":"openstack%2Ftrove~master~I1eb95f94ca1d242e4539b4a0cf5906ee62f74284","project":"openstack/trove","branch":"master","topic":"public-db-4-release-note","hashtags":[],"change_id":"I1eb95f94ca1d242e4539b4a0cf5906ee62f74284","subject":"Release note for public instance","status":"MERGED","created":"2019-09-11 07:27:17.000000000","updated":"2019-09-11 09:41:53.000000000","submitted":"2019-09-11 09:39:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681393-1568194751527-1839ade1","meta_rev_id":"63d680a4108fc8b07488997159f5cd7d33cca3db","_number":681393,"virtual_id_number":681393,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-11 09:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-11 09:12:03.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-11 09:12:06.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 09:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b381ea920c621efa3dfb9ac30b73a72024d7398a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 07:27:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e235c74a206683e4bad0265ad83a192b7af3a63e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 07:27:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"047bb20eea6b1d6c340699aaaf63f94a8e7923ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 07:40:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b3e560d5a84dfe8d5770a7abb5db4a : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b2800c651b433a9a265c81fd60d2db : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1bdc563267148d59b7be04fa2be3197 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2d694eb48c494899178746539a3371ff2a300a0c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 08:45:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a1102a7c9d7885814731526504b4704c3dd4fee6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 08:55:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ee1503165c417b9ddf8cef663c1bc5 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5f335a736b4d599d611b48c37bd369 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1567cac1ed454841ac47ade83ec55911 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9a737e57cebb851fc4a862648c60c52db69412b4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 09:12:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"39808dfecff420ec9b55fc899f87cc1dda238a15","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-11 09:12:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b2bc1dce7cc3c461c0a10c0009c6abfb27a8378f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:12:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00af0b60a524994b9910d0e655f62ec2095c6323","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:39:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c13f1c1d4d1821f89a64ec0c778d9c917b8bb590","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:39:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98af40077bb04a7d95ade9816f40f732 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c5c57bb7549aab5c039ad133101cb : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57be08657ade4ee0a623b2b49ea9c567 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c245d010396bae2c52381127050f9fd5d323d605","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:41:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6990ec3c6143c8a61970237ba70cf8 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3310e3ce79d8497e9788e9dd56797089 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19d192e55fbfe6d8fba4cd6a2abdceda8a4ca9b5","revisions":{"6958ec6e5bff97a6251cc0a3eba89d8bb34aab0a":{"kind":"REWORK","_number":1,"created":"2019-09-11 07:27:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/93/681393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/681393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/681393/1"}}},"commit":{"parents":[{"commit":"b04228d4e8e3d4bb7f4b031054009badd754bdfa","subject":"API doc: \u0027access\u0027 support for creating trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b04228d4e8e3d4bb7f4b031054009badd754bdfa"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:27:02.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:27:02.000000000","tz":720},"subject":"Release note for public instance","message":"Release note for public instance\n\nChange-Id: I1eb95f94ca1d242e4539b4a0cf5906ee62f74284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6958ec6e5bff97a6251cc0a3eba89d8bb34aab0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6958ec6e5bff97a6251cc0a3eba89d8bb34aab0a"}]},"branch":"refs/heads/master"},"eef73cab3726b6b8f1f653e512ec9ae81a2566fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-11 07:27:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/93/681393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/681393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/681393/2"}}},"commit":{"parents":[{"commit":"b04228d4e8e3d4bb7f4b031054009badd754bdfa","subject":"API doc: \u0027access\u0027 support for creating trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b04228d4e8e3d4bb7f4b031054009badd754bdfa"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:27:02.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:27:23.000000000","tz":720},"subject":"Release note for public instance","message":"Release note for public instance\n\nStory: 2006500\nTask: 36467\nChange-Id: I1eb95f94ca1d242e4539b4a0cf5906ee62f74284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eef73cab3726b6b8f1f653e512ec9ae81a2566fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eef73cab3726b6b8f1f653e512ec9ae81a2566fa"}]},"branch":"refs/heads/master"},"19d192e55fbfe6d8fba4cd6a2abdceda8a4ca9b5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-11 08:45:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/93/681393/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/681393/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/681393/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/681393/3"}}},"commit":{"parents":[{"commit":"94dd0bd2b7cc57d69a3ba23345445d8b115600db","subject":"API doc: \u0027access\u0027 support for creating trove instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94dd0bd2b7cc57d69a3ba23345445d8b115600db"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 07:27:02.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-11 08:45:33.000000000","tz":720},"subject":"Release note for public instance","message":"Release note for public instance\n\nStory: 2006500\nTask: 36467\nChange-Id: I1eb95f94ca1d242e4539b4a0cf5906ee62f74284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19d192e55fbfe6d8fba4cd6a2abdceda8a4ca9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19d192e55fbfe6d8fba4cd6a2abdceda8a4ca9b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
