)]}'
{"id":"openstack%2Ftrove~681996","triplet_id":"openstack%2Ftrove~master~Ic141900cd5c2c8ee0185af72cf81678cf3a3afe3","project":"openstack/trove","branch":"master","topic":"support_interval_check","hashtags":[],"change_id":"Ic141900cd5c2c8ee0185af72cf81678cf3a3afe3","subject":"Allow change periodic interval in database update","status":"ABANDONED","created":"2019-09-13 08:54:31.000000000","updated":"2026-09-17 17:23:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcab1e4ff1b1d878dfce6ac2385cb232149cb1ff","_number":681996,"virtual_id_number":681996,"owner":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-17 12:31:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":-1,"date":"2019-11-27 22:24:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 12:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 22:24:09.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"7ff490a14161fed575d4421f0b22849dd967ad01","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-09-13 08:54:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9234bc4db0e519f2ee44237c7bee05c502a30d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:32:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/460f366cbc124b1b929293e359657968 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e512b918007149f08049a594398db18e : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348a702fc4e649379fdef2cdcb023eea : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/593e71bdeb9243f8b97c32e384b6a0e4 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76172346b8644f869ece755783dff788 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76fbdd2c717d43d3b38f428ff1464c18 : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1f434a53b24539ba88cb6e0dc3947d : SUCCESS in 8m 36s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c30efd2b5aa04245bd1bd9c8242c15c4 : SUCCESS in 2h 31m 33s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c3c3173cf75540ee80cc71819d846da7 : FAILURE in 1h 16m 41s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b3f06cc603864a0498c9dc6f1c77100a : FAILURE in 1h 01m 32s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7e54aed9cb2b44c4be5ce58595e83680 : SUCCESS in 1h 47m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/bc066f0c99b44a2f8280ac53b5ee1665 : SUCCESS in 1h 44m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8c1bd7609cc44e1cbd4a81f70000defe : FAILURE in 46m 59s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/bd11caf079e047f1be08ced0aff6ed19 : FAILURE in 39m 41s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/ece171bf4c424b79af51134eb118e0a3 : FAILURE in 42m 45s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/f355c7bcb8d1402981f32473bd8a6e86 : FAILURE in 46m 58s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0534a730b08d4def9f410be388bc527f : FAILURE in 52m 13s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c8eb43bb597045799cac1050507a6732 : FAILURE in 40m 38s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c6c26eb42c5f414aaf1e23416c185afe : SUCCESS in 5m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a652780de1b4adba722bcd9786d7f45 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a082a10c5874f8ca8bd3746da551a7a : SUCCESS in 7m 49s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/613bcde3e4074446a612e4501506d034 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":1},{"id":"77c3af571d9b87c85e2b024f2df240d3a83727a4","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-10-17 10:08:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7e8813706eef2d5575c305e7444954a81cf6a1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:31:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac8f17609514b65b7ee9f3a0b4a4660 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92fe7c7c4d55419999b767497129c903 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77caa41437dc4b57aadc10dad46ea28c : SUCCESS in 9m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2ee5fe7af6740ff8e7067b0f680cdbb : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4d7ee8f2b649dabf818149818af756 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5b1a3e52fe24fdd929084854ed81f24 : SUCCESS in 13m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e132c52b7dff4e5b8b85e11cc8c89a63 : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b82190879784452faabc1521bf05fcc6 : SUCCESS in 8m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/828df825e38f4e249634b796c20fe5d7 : SUCCESS in 5m 42s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/15be77aa59dd4cdaa3f30bcd4a6d31c2 : SUCCESS in 5m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/503266824893498abae8c9d2b4ceba68 : SUCCESS in 2h 18m 03s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/17c8941f6ddd49d6aaec69495161f6fc : SUCCESS in 1h 51m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d935ceccf6884b0aae059db0515d5683 : SUCCESS in 1h 32m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d67831d7f94d4cc8b050086f0534f475 : FAILURE in 45m 23s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/708d89cdc831449c84d6482c034a28b0 : FAILURE in 43m 43s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/39962ac8be5047c5916d3c4374ba7760 : FAILURE in 1h 19m 22s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e1c6b41b576d41899752c1ff7d9940ad : FAILURE in 1h 01m 49s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/7d74f9e473e144599ebab22299f9df15 : FAILURE in 44m 49s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/87eb68b9c1b54f8587388f498afe4f19 : FAILURE in 37m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01be553045834d72913368d531b09abb : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f708f6a2b92b8e5e83af12bbf2897f72993535b1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-11-27 22:24:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nAs this patch adds a new config option, please also provide a release note.","accounts_in_message":[],"_revision_number":2},{"id":"bcab1e4ff1b1d878dfce6ac2385cb232149cb1ff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:23:58.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f1b0fb0a909a7e644fa0d54a11de8d50102385b","revisions":{"3d77f8c8222b055cdb6f3a2ffd962a145890cd18":{"kind":"REWORK","_number":1,"created":"2019-09-13 08:54:31.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/96/681996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/681996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/681996/1"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-13 08:50:11.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-13 08:50:11.000000000","tz":420},"subject":"Allow change periodic interval in database update","message":"Allow change periodic interval in database update\n\nOperator/ developer, they can change this parameter\nfor guest-agent.service. Depend on networks and\ninfrastructure to change suitable value.\n\nChange-Id: Ic141900cd5c2c8ee0185af72cf81678cf3a3afe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d77f8c8222b055cdb6f3a2ffd962a145890cd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d77f8c8222b055cdb6f3a2ffd962a145890cd18"}]},"branch":"refs/heads/master"},"6f1b0fb0a909a7e644fa0d54a11de8d50102385b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-17 10:08:03.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/96/681996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/681996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/681996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/681996/2"}}},"commit":{"parents":[{"commit":"bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48","subject":"Merge \"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-13 08:50:11.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-10-17 10:08:03.000000000","tz":0},"subject":"Allow change periodic interval in database update","message":"Allow change periodic interval in database update\n\nOperator/ developer, they can change this parameter\nfor guest-agent.service. Depend on networks and\ninfrastructure to change suitable value.\n\nChange-Id: Ic141900cd5c2c8ee0185af72cf81678cf3a3afe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f1b0fb0a909a7e644fa0d54a11de8d50102385b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f1b0fb0a909a7e644fa0d54a11de8d50102385b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
