)]}'
{"id":"openstack%2Ftrove~682261","triplet_id":"openstack%2Ftrove~master~I3d909082fb887b24ed25c93e11792031d5053b1e","project":"openstack/trove","branch":"master","topic":"image-job","hashtags":[],"change_id":"I3d909082fb887b24ed25c93e11792031d5053b1e","subject":"Add image build experimental CI job","status":"MERGED","created":"2019-09-16 01:19:37.000000000","updated":"2019-09-19 13:59:42.000000000","submitted":"2019-09-19 13:51:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682261-1568901118947-fa3c3f25","meta_rev_id":"a4a35aaaf9f1cee26012a88398ebf6ab6d123e34","_number":682261,"virtual_id_number":682261,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-19 13:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 07:25:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-18 07:25:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 13:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a595bf3686533683ab55f767e843d290b52f2d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 01:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f739a9703fad6df5be03805b5e1745bceb6e7d0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 02:20:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3afb8a111864818a0098a6c59a4f9d6 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/001ac6ec650d417d9c677e4a1cfb461a : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765044b77f9942119c84ac131e4fbdda : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df3cf7d6ff4b44219d5f46d848412ca5 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d14ad848cd5435ab86161288d8657da : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68ae6907fa5244a1ac04ff23d675770b : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e21955117184d7281bbeca2115c0c84 : SUCCESS in 4m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8adf5b33cd854a9ca8ff8eb1182e3df6 : FAILURE in 24m 49s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fd3c4aa231614da99d821d23ef32b140 : FAILURE in 31m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7a77e247bbea4191bb1a90b47c116154 : FAILURE in 29m 30s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e25777b86cbc44069e1cb3199d45a682 : FAILURE in 31m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a55f5e05fa2e4f459dbe148441bd726e : FAILURE in 55m 04s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/0a3ae2a76ab44c068e8d6d6148206f78 : FAILURE in 53m 59s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/accf318ec8be4b45b825dfae724214c6 : FAILURE in 54m 22s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/6bbd5715fc8e43b7bab7b9370c6b72a1 : FAILURE in 46m 54s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/d1e69b4750d24af19233e403f6026aa6 : FAILURE in 40m 39s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c99b4c16ae9249efb2e028372f5ce214 : FAILURE in 30m 32s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/0ef57866befd4b46af41af7a3f8a0501 : FAILURE in 31m 27s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d4d66aaf6219447d9446b127f7ae48b5 : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0727e553220446a1a49f36e16f2c6786 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ec293dff50a4e60a1d9b36f2fdaff84 : SUCCESS in 8m 14s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39c81bdf4f64447c8939548ea7bb2de7 : FAILURE in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3f6253e1fc335a2bd26f789cfa6f0d7a19fd4617","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 10:41:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ccc93676436bdfe2ac67f1c42d5afb99245132c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 11:35:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/179c42a2e49f457d895e35f91820bc72 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39f3461a6bf84739885f4b64299bcd4c : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c773dca21d42c0821d58ea607af4d9 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0039ffa11b6f40919441ce4bfe2c628f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5b2492f26f42c2b0da43faff637a89 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ace3660fe7a4046948b3ced5bb2018b : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de506d799d0940909777fb680d0c15b8 : SUCCESS in 5m 43s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ad9fad6062684d38a1ece3a1791d4e15 : FAILURE in 42m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/07e449e687c54a4da555362c93cf13a2 : FAILURE in 40m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/67c18fa6624c46d2a7640c7da6f82774 : FAILURE in 35m 44s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a7a3429062f24a14840c88c730549946 : FAILURE in 36m 56s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f16e71173f79425ba6d4a022938e5172 : FAILURE in 39m 40s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/bc8db0f8ec9548e38c1dc7b691d4549e : FAILURE in 45m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/1e4867df292649d58ed9488ae3baae44 : FAILURE in 48m 50s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/3a796302003b4a8290f3014ae53f7c47 : FAILURE in 34m 39s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/2c5cf92acc6c4e8c8aaea7f832bb0dd0 : FAILURE in 35m 47s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c363bf2491fd4e2ca87fcc44cbbe7138 : FAILURE in 37m 35s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/7695b494a6a34fba882853a950eb8c28 : FAILURE in 42m 58s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6c92e592688a48989e3b0b27e75ed33d : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a0093a23f8e4e13aab8dd589be5de25 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9677ac0f96b54615a3de95e1d87ba620 : SUCCESS in 9m 14s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6197522725b414fbf0c7038de9adda6 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e6948a3b09c5238fe81840b003afe93f18d35f34","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 11:46:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"26c180bc92ed194e6ebe9c2552946e8a2fcdcd4d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 11:48:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1a9436baa741209115dd587e2fa21b24a82b1850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 14:30:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/804f0a6d1a8c415894fbe79090a816fe : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d66ea451ab247009c742f87b314d5fc : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026a76302bee40bcb6c90266fad2cf22 : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c038d03f79a3491181b06c70d7d6ebba : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90baf6b9f22d4b66981e0e79f6dbc8d8 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/240d20119b3a4662894600170786c83d : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab8f24a15b647b6a314c83111270c1f : SUCCESS in 6m 01s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d535cfb137eb4e58a9daf93da32dad12 : SUCCESS in 2h 34m 26s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d4cb5f61cecf43efbb49577e91a9aa88 : FAILURE in 1h 24m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f4e89c99d69f4b42a392b136b3e8b05f : FAILURE in 52m 15s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3fd92a4ca6964389879955b63b4fdb12 : SUCCESS in 1h 37m 27s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fec67ccf20204badbae7575058ab382d : SUCCESS in 1h 32m 00s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7842b7ad24ed458696294837654a25ad : FAILURE in 54m 08s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c31a318fdde64a5c9b961c6ca18c5de6 : FAILURE in 51m 23s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/21c5fde0171b4221af0454c0cad3a245 : FAILURE in 42m 37s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/8db88836578c492084c8cc49c8b80c23 : FAILURE in 55m 34s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/eeb313152c864633ac4b56952f5dae77 : FAILURE in 47m 03s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/722d5bf2d3dc40d6992fea9d085cb398 : FAILURE in 47m 16s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/3d738817b8304ce388c9afe3efab69db : SUCCESS in 6m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d5bf90bb893243e1b228543a833adb01 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb0eb2bb0969470f8a3627cb168ff798 : SUCCESS in 10m 18s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955a8805f60b4bfb946d8f9fe384a3a3 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":4},{"id":"0035a3f3cf253c5ed17b7de10f6e7edb72510818","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:31:25.000000000","message":"Patch Set 4:\n\ncheck publish-trove-guest-image-mysql-ubuntu-xenial","accounts_in_message":[],"_revision_number":4},{"id":"af9b418cdf43a9aceee9783dd29c962e68694aa8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:31:58.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"a2cbe76bb5a3848d5760476731506af96168c445","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:32:14.000000000","message":"Patch Set 4:\n\nUnable to freeze job graph: Pre-review pipeline experimental does not allow post-review job publish-trove-guest-image-mysql-ubuntu-xenial","accounts_in_message":[],"_revision_number":4},{"id":"64e09516dad204a8c2e813cc1e319d57af26fa73","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:37:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a0968ca5f42ae3108c1d92cdd7eaeaf08b40c444","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:38:01.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"8c7267573661f12bb50789b29542bf24374a60fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:38:16.000000000","message":"Patch Set 4:\n\nUnable to freeze job graph: Pre-review pipeline experimental does not allow post-review job publish-trove-guest-image-mysql-ubuntu-xenial","accounts_in_message":[],"_revision_number":4},{"id":"25984ba1bfaf3671dccfac8145f2f56237291510","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:42:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"facee5eeab2f0698395d3387806c2a5630ad6a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:42:34.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  Once set, the post-review attribute may not be unset\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-trove-guest-image-mysql-ubuntu-xenial\n      parent: publish-trove-guest-image\n      post-review: false\n      description: |\n        Build and publish Ubuntu Xenial based Trove guest image to\n        tarballs.openstack.org.\n      vars:\n          datastore_type: mysql\n          guest_os: ubuntu\n          guest_os_release: xenial\n          guest_username: ubuntu\n\n  in \"openstack/trove/.zuul.yaml@master\", line 356, column 3","accounts_in_message":[],"_revision_number":5},{"id":"69d7d079d393fdecc927daed453b3ddb6686fe7f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 19:46:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d64663f4203f99e64f56a2e198627efd322694e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 22:20:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e58ef84b7bf4d09a3e3136184b9fb2a : FAILURE in 7m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f65143a5a2420b8979da0523faedc8 : FAILURE in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785ec754746f4cb3b21ab960440216d8 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b9ed91dd80149d29cfdae9467203312 : FAILURE in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00006a12e8074210a40d3ba5638c3614 : FAILURE in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7a46f58eaf14540be9cd96e25ea2482 : FAILURE in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ea794c215344fb98caa4830da5a53d : SUCCESS in 5m 40s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/72709ac62e20413796934c939d138745 : SUCCESS in 2h 26m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/305fd71652354f9692f7c9dbb4ad99fb : FAILURE in 1h 17m 36s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4c7c5606eb2846da8e32afa303c203d4 : FAILURE in 1h 02m 10s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/33b0501873f94233bea2d1bafc602b53 : SUCCESS in 1h 46m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d75ff8ad106a4065a17204989190e65d : SUCCESS in 2h 01m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/189487ec6d9d4696a9a6332036d9b1a4 : FAILURE in 43m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d70e8a39f543437b906edba98fcafc8a : FAILURE in 48m 28s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/d78773677e634ac99954318656df18e7 : FAILURE in 42m 06s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/f0eb291bbc964704b84cf34167f6fbc7 : FAILURE in 46m 32s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/cecf9d5b721e48c889fa71c0d1900797 : FAILURE in 47m 02s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c5aec202b5164fc7babe38bdea323eb7 : FAILURE in 42m 50s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/1349f8b7d1dc41f086803b5f0771a8cc : SUCCESS in 6m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bdf12dc5ee664183be1bf0b1c0c1243d : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f293885343442a797306a2ee39f29bc : FAILURE in 9m 30s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbd3b22847b64ed680ad5dda1c903e42 : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":6},{"id":"0866aa2adcda92c402feb201f6171e3653b1d923","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 23:13:42.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"63c4549dd02f85f9ff15305c29f1e6fb182461a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 02:12:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a0964f312754d8883a87c0441741894 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63818280f99c4eeca02c5379f944aea9 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce3e80e5bd549a6b335fbd48c816dc5 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce243a2775964dd5b0fa94af7fb23cc5 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7f18db90394ef1966b7d3dcfb9f448 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/171297935b2f4fe79d33c3f78c24c0a7 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a312e1c3a44a4060a766b547aa3cacd9 : SUCCESS in 5m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4b30772d17934ea0a5f1c0a7cb26757c : SUCCESS in 2h 27m 12s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b05b9008b52841549f6230ee6e5af928 : FAILURE in 1h 19m 49s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ba8e2516fa56493da19ed3abffef6cec : FAILURE in 56m 48s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7dace3989c3247f3be89f1bced634e81 : SUCCESS in 1h 46m 13s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2388674963384f43ac5a6e598d48e5c2 : SUCCESS in 1h 32m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/029bd6f0e4d443d2a3104bde2500794e : FAILURE in 48m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/de1d9859afe84b6f8cc4a28fd5da9bae : FAILURE in 42m 35s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/fac8fd359f8c44ecaf933143a1f945c1 : FAILURE in 46m 17s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/cf1794e7dc2e46b5b112aff1f865c4e5 : FAILURE in 1h 18m 30s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/db25796397b04eabb207ad21416ae993 : FAILURE in 43m 16s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/54414552d9724d19802a28b269b4bd3a : FAILURE in 50m 19s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/49353f2ca9fe4589a1fe8dfb6ab60ef1 : SUCCESS in 5m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f87613a574f4842a314076a0bdac5d7 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c0e9a42deff4db897d570ac367d5d1c : SUCCESS in 10m 00s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d93681fcfd74ac89e4b1ec44b3af294 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":7},{"id":"3b45faa33065461d98a469c28e9dc02668334433","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 02:13:17.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"fa084bcaff183e4b7fd07203bb76e18d2f4db6b6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 02:13:29.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"b853201e708bfc51a89a459b76d6f63e1cdba35b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 02:13:44.000000000","message":"Patch Set 7:\n\nUnable to freeze job graph: Pre-review pipeline experimental does not allow post-review job publish-trove-guest-image-mysql-ubuntu-xenial","accounts_in_message":[],"_revision_number":7},{"id":"4cceaded8fae011fca98e600896ec25f8950b439","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 03:10:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"74c0f80233f541860d83d404fe7099fcf6f5b937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 03:11:08.000000000","message":"Patch Set 8: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-trove-guest-image-mysql-ubuntu-xenial","accounts_in_message":[],"_revision_number":8},{"id":"f381bffe7ae9373a43a0262fa9adf1d662984199","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 03:16:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1ea0b4d67aa8fa13f4fe7f7f202255589fe4b48e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:49:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa426ba66f704b65a7c5ed73a079ae30 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4930a7b69c39449f918aa2844a77e516 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c75e97f4e64324b39446397de5bb70 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fc239dc968447869640a62f123bcbb9 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/136cf54920f9494b83673e5cb7571902 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/664beed4e5ba45688ba45e29cfe42670 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0887b751856a41fd8be0b7768c7eb07b : SUCCESS in 4m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0e3d6b0e94640c19a1bd82aa78e9786 : SUCCESS in 8m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/294db8bd83f3441c92103aa5b7e8381e : SUCCESS in 4m 59s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2e0aa4dcc0af45fbbe3d0a89184da07e : SUCCESS in 5m 00s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ef11ef1731f54868819e90954d68ab61 : SUCCESS in 2h 26m 57s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/320d9a6b8f0544cfb7ab9b3a50b10f67 : SUCCESS in 1h 43m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f6098deb9e084e5daeee46603d6815a7 : SUCCESS in 1h 49m 21s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/71327d3ce9384c2d8d1ea25b196b3830 : FAILURE in 40m 36s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9e29a0a0494d41f2aaf34c329d3e8619 : FAILURE in 42m 31s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/dc07128337174713adc602c104e54316 : FAILURE in 1h 14m 46s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6dadeb124fa0460f8f5aae0d73617691 : FAILURE in 54m 45s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/eb6e7caef1a84fa89763732451cfd77d : FAILURE in 45m 01s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/1d59f7d1cd774f21a5ca99ee67f6dd42 : FAILURE in 42m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c51680c49d94404943310afffb847b2 : SUCCESS in 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"171aea0349ad32008725248d7df9c166c34fcbfa","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 05:50:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c801d95bacf46fda4d79c6ea798b49f0332da0ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:50:17.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6bf3ced34710fd373a061bcddb0939e65a70e15d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:22:44.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8b6735b31c4a6dbd1c27ee9f2e4fcf : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b35549e342284322b19dca298ab20db2 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17cca2519ae3406abc0c9a605c0102b1 : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc17dec900724bbb990b23e26cb09ed9 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/235c25d654004bed9db63ea8e36a36c5 : SUCCESS in 14m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3bb22eda444d268514e12b61427589 : SUCCESS in 8m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6277c5dd7bcc480a8229566f0f07e0c7 : SUCCESS in 10m 01s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/31a788689cdf48fea61f5cee83171016 : SUCCESS in 8m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e97705d8a2a74f9ea4da3e85e52f8dd4 : SUCCESS in 2h 27m 45s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1945de388ec94805a08a1ff8a2734442 : SUCCESS in 1h 46m 26s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5bd5f74f4bc747bdb6b23e198461b7aa : FAILURE in 1h 37m 28s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1380d1b9e084206a8427967296d4f01 : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9195e7b3d4f33bed3559e11b631ce18f48d5672d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 10:23:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"53f124e66b0d082ca9d69e162bd400000d7e1ddb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 10:23:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2a85a05e14d1e55ecd2026984c0b73b5013f9f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:13:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab46158c21934067b35ff6f1d5594770 : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da1c2dfffc2445e29233e1a5e219faf2 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b6253c86ae4534a038c6d4e9b0e166 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/215ece86f58c4b668b6203fb4644f4b2 : RETRY_LIMIT in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0910492ee91c469badaa554c1e669aec : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de3e8cedcd77435bbcf181acdbe31ebd : RETRY_LIMIT in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9191ce2a97af40cd987de0becb1bbc0c : SUCCESS in 8m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21c9130c3aaa49ad9072fdf4e3086d45 : SUCCESS in 9m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce687676714c47edb7ea42d4a12b0fc6 : SUCCESS in 8m 18s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/4d575ed702c94c619c44094a3ffb7856 : SUCCESS in 7m 39s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/bf3a36789a0d454583a6dd3abef978fa : FAILURE in 2h 37m 04s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dea0225ab83e411aaa3ef5f3496c55c3 : FAILURE in 1h 39m 20s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c8e69e7a752c46bfbb8035e2645c8924 : FAILURE in 1h 34m 10s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/abed1d2bfde044cfa9d9a1780d6a5909 : FAILURE in 47m 16s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/81f2e142b7c24357ab65740c856e04ba : FAILURE in 47m 06s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/890a14e33c2b427890dd488d98bd5f68 : FAILURE in 1h 20m 21s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/eec250db697c41268b7fba796327c091 : FAILURE in 1h 01m 02s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/b640e5dd26a74e3f9a1818839945b3a0 : FAILURE in 49m 28s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/d2ef0ac77cc9425083d15cb927b958d2 : FAILURE in 43m 51s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8d33b0c02624fe3b2addba4037bb098 : SUCCESS in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bbd196ad643ac4cf207b52b9a1220328e8836856","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 21:56:30.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"cc6c3efc1283765c76cb8ddd9245115f04533731","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-18 00:03:00.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"8768561233d24584317fc2c4514f0981ada81e1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 00:34:54.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1be7524362444aba96693e6321344013 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/960b445089d94a2fac225c4a4404da1b : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915d087695994630a18e8bfc4c63139e : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75c805c37df3400f82de9817aa701fe2 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7b7cbbd7ef43138f38b43d3b320405 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5632dc3824f34e6aa24089a0b6d3ab7a : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0f8d50ab704d7483b678d7fd950e86 : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee1f719fc1d3419fac064b9830d44bbb : SUCCESS in 7m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40cd96d9e2034a3b9aa38c760ccfcd60 : SUCCESS in 5m 02s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/298ad3789c964ea99c9638f8e5c273a6 : SUCCESS in 5m 33s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/debba7b535164dabb1a70b9f6eb8a908 : SUCCESS in 2h 31m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a97b61d755f245368ad54f69ef7af7c2 : FAILURE in 1h 45m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0d808d7fb4564b4d83baacc22de6c851 : SUCCESS in 1h 53m 58s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4b2dc575e2b249258dc092b5b9f42edb : FAILURE in 1h 09m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/16850bec586b4323a448ecd1ab107444 : FAILURE in 38m 26s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d9f6101b9b7b43738024b0db1c356f98 : FAILURE in 1h 20m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6e04da7ef39a41589bf61abb67f89fb1 : FAILURE in 53m 20s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/20744f5069a3484f9cb76b4e25f13680 : FAILURE in 39m 56s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/4515d1a9bc244959b33551facc1451a1 : FAILURE in 44m 38s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/480bc97804b24ecf96a03b0676f87569 : SUCCESS in 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3206d05b61b4596b464ff802e4e3f29d802f29c4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-18 03:52:11.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c09636bef92fa0bad8d07a6acfeada977c97b868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 06:53:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7982329e0614db3967aacca7bb38d5c : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9607236d2ffa4de39eb2a4adf61de42e : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7228e79a9c8a45b685740251603bb18a : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf7bd868532b4484a6b66d4f840f48ae : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50ba78ed9f5a4af7a9c8fa7ed7460492 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be07852b7f9147fc8d36778174bcfc01 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8848869bb3a147999e1fc7363886dd00 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58b2929838424aa4b41788fe6226f400 : SUCCESS in 9m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc7de0b034d4482ba6832b19b1b55709 : SUCCESS in 5m 31s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6a6abe96217a4f3599c514dd85a1a887 : SUCCESS in 4m 46s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/09e4e27d5d9b423b82302362c846b637 : SUCCESS in 2h 54m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/52aab60bb7924ae5835cb6c923a590b1 : SUCCESS in 1h 44m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fd5ac5b82db949ac87ce3bb18ac881b1 : SUCCESS in 1h 44m 20s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/172882196ad54a319095bf936a7bf11b : FAILURE in 43m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/a0241bef048b4f24a5161b444a37fd50 : FAILURE in 44m 17s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d14b77ead9ce41978a3bfdd7f5f54ef6 : FAILURE in 1h 15m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/edc7fcddd4b44237aff877062b1f9972 : FAILURE in 57m 23s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/2abf4fe030bf4a1da2acbb94542a0e87 : FAILURE in 45m 22s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6dbd217f8b694f4b8bb04e62e57936af : FAILURE in 41m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a354aa991304b92a50d8d5e39462276 : SUCCESS in 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0758a8fa6b39f3db296c386e1382e2bc4ec0fcf0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-18 07:25:55.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"649ca499b3c6774bed1904cc0d32c65f16076fb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:26:10.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"19d235cc3592511666c0a2c6f18744f9c3f06006","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:56:29.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b3ee414028b47a9abf96e08a4f2e71e : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbfde596619429facdfc74e5619b380 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fad5b1679614c91b92c1b1b22898579 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b4d2f6716b4527813ff8657ad7844e : SUCCESS in 9m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8c997c25b8e447590be48b135acbd93 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff3cb9e954549538834d6c4c3308b9f : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f33bf5dd7eae45cbadf56dd336aaf883 : SUCCESS in 9m 12s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/115deeee066f4b5e9345ae3029d9f45b : SUCCESS in 7m 05s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8c83e95f644e4242bfbd54c4f4d8214f : FAILURE in 2h 26m 39s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fc7d2aab448b43389e3f3ba1bd69621e : SUCCESS in 1h 43m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4894e7acb8ce448ebb0fd5857dc5473c : SUCCESS in 1h 46m 00s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53004439d79146848e232cbe6d66e05f : SUCCESS in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8833364b5e7ff08a7d0c1c80d745d1f955af2203","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-18 09:56:58.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"9d9e16c04383c0e6e34dacfa4aacd58c93b58ce1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:43:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ec9d3d6250411a9ac27d5bc9942272 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e1a0d232e174b46a2bf873b549b1dc4 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05847cc3817441582bbbeefab427fee : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b1929e38da242d58fee447dbeeeafb7 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df7382a4a72475fa5122a7924029f87 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce960885021a4186ba1c493778447c4b : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e96bb1e76840e39373b4450cf3ed23 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54ffcb527b1a48f7a7c840d5af60b935 : SUCCESS in 7m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e34b463dc32f493ca5ae5fce2cf91b71 : SUCCESS in 6m 36s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c0eefd03a5ee4a64aa5a15c2159d19ea : SUCCESS in 4m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b1efc69f52184fba9441c3cd4f52e3b9 : SUCCESS in 2h 39m 15s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5483056c0076448aa365caaf9c81ec5d : SUCCESS in 1h 48m 20s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/324932687a434d0299ddfa364101f881 : SUCCESS in 1h 56m 41s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8ae6b1b2fcc14db891ae2fcb4102bf0f : FAILURE in 49m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9a893a88d2d94f5f9c79960cd26b4123 : FAILURE in 50m 31s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f0d06718310d401ea1c9577aeb50a061 : FAILURE in 1h 41m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/36be7fff948b4fbd8477ee895cdc35c0 : FAILURE in 1h 04m 28s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/99a70e23723a41a6875de6527614c502 : FAILURE in 43m 36s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/4802ea704e0448fba7ff8cd14d0fb16f : FAILURE in 49m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3c421d8ad741c4bf124e61933e01e8 : SUCCESS in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"253ba2e91a6638694db20ff899b27b8300f4acd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:44:07.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"7ae41d878c72be5302a8b4c60100c923c4de5a36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 15:21:34.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/034654a7c60b4954ade99e8ebc8ab79a : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2322838f7f84eed894536b7f0e471e4 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bc0baa55a5d41c8bf35b00ebcec027d : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/352f91fe3c624cafa30d306efaa52b31 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aba5357e8daa49b1bf01bf7d41527611 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6018591757249c1a5e3e962cd722166 : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39ac79cf3e1e4471b850561a53817c07 : SUCCESS in 8m 28s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/1c78f006c46e452abc3af945664efaf9 : SUCCESS in 4m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b812b89297cd4bc8a7b36ed6eba18f71 : SUCCESS in 2h 31m 26s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e2176de26d5b425fa907028311d992b9 : SUCCESS in 1h 49m 11s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0e6c8fa3756b407f846b22aeb466a7bf : FAILURE in 55m 56s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80fd888d8d9f4f57a057143aff430fb8 : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"77d6a52d3943c62ec7453189bc4245e5e29d738d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-18 20:09:19.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a4070fa831455500c8d64cb988d70d8c13f148d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:46:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b82ec3bcb424044b8cc3515bda66c4e : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39470afca2ea43a4a4391c6d670cafcd : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946b26729ad747b5a3a1d854933cec7a : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eed23cba07dc40a6aa04f1158dbb3356 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65b8a6e630884f1097792c1bc73fc3f3 : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e57d1a74be80418594a35391d8f6c724 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a229757b4b949d19f94e5c2489bb08a : SUCCESS in 6m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ced4701e8db4de9afb01c2094cfd2fd : SUCCESS in 9m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f84ff5bc3fa6471287f6affed0d067fc : SUCCESS in 7m 24s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/fb5e36e293dc4097bbd359c7859e2511 : SUCCESS in 8m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b2bfa0cc11334c738e1b61bb7bd82ec0 : SUCCESS in 2h 35m 59s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1463f8ddc9fc4a739afee1e514a5b856 : SUCCESS in 1h 49m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/63052fa253444bf295cac80e3c112526 : SUCCESS in 1h 45m 06s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a960db2f941a43ffaadeba2b7a80b979 : FAILURE in 46m 32s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/959a970a0af34ef1b33f9ec6bf521cad : FAILURE in 45m 51s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ee8c05df066b433caa5d6a5d1fbf4fed : FAILURE in 1h 21m 56s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3178a5932be94dc4b669923db6812637 : FAILURE in 57m 07s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/030f1f7afbaa4a588b0737d9b917403b : FAILURE in 48m 24s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/5f61fad3425e46dcba688fe177a38535 : FAILURE in 48m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/249be9fcb2f54b6189ce3cf91a1f53e6 : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"24f01a746654df6ea247673537322eaa7c8135b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:46:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1ec5313982681654a2c5e1e781cfda8c8211ec80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:02:34.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85246bd6a7f447d69e533e61f326cf6c : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68fc4d63d154ef182ec877836bb34ef : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f723401289fd4c0b957a527157c2e1ee : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00f80438cb1740e183bd6fa56cfe09cf : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a6a7b2fe7774f2cae7382b4185830f8 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcde5d030cbf49798a9d8ed8d3a39d55 : SUCCESS in 6m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7711eda0d49b48df86e039ff751eeaf2 : SUCCESS in 10m 43s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/5f8a7facafec4cfead623df0945c7681 : SUCCESS in 7m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1d4ed3f007f54b578c21f8a5e798924c : FAILURE in 1h 48m 09s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8127bed772ee4742957371e2ef332720 : SUCCESS in 1h 44m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/8d39acb0ec684182ac14c732b296500a : SUCCESS in 1h 53m 41s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a23c32ba01fd43088b68a52079ea6965 : SUCCESS in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a294c1dba1815bf7c9a6873c068c70899e83503d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-19 03:05:38.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"2f2c7df03317dcf170d495c1b9b987cbcc29ea65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 05:41:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b7c733afba94637ac19304705baed74 : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8deaba12de284a7890cecd65145c962c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a39ec066974ce6b251153bf2168a0a : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea3f43661f05439e99574f100ad22da6 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85eeb445b8b04527a21d73eacdbdc2c4 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe2928190c354c8ba7032321603ce837 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc52a86c62e74f55b7dbfb4f57a8f04a : SUCCESS in 10m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e7d3ce277cc45038769486d0a0fd3ec : SUCCESS in 9m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc706f3b78454ae0aa14166777e92879 : SUCCESS in 6m 03s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/4aa64738712e49119a2a64ddd6dc5da5 : SUCCESS in 7m 36s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0abed2c8781e45249256555c8ecbecc4 : SUCCESS in 2h 28m 02s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/15e747da22c849bc9f3508eefc0f38a9 : FAILURE in 1h 57m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/28c977f1d65a4402bbdcaee360617bf4 : SUCCESS in 1h 46m 17s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/385df9b827c04218a7cbe3fbb28c05e4 : FAILURE in 33m 51s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/dcf55657df5f418db4189d88cc35fa64 : FAILURE in 42m 46s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5536ac55beff46d4a06cf64d07ad5305 : FAILURE in 1h 39m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/54df25e60ad545a8a9380c1a889cfa54 : FAILURE in 1h 15m 28s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c5913d1143de433cb8a5798b3051cb5d : FAILURE in 44m 33s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/4790ae32f23b49ad8d5dae2ab72243dc : FAILURE in 46m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/883889466d8f484da715edbb11578681 : SUCCESS in 54m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"099b3ce73c71fd014366ef2f8ec94bf1122f5f94","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-19 05:43:09.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f845ab859db52f62de330b69bb4322ad97221a02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:12:25.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02df15b3af68434e8140a43228f31450 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecba3cdded214ec29bffbaf7ed3eb6a5 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885b51050d06435a989b544b53a48c23 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11916a98bc9b43e7af56b9542be5722d : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b39e8815e1a14b3a845ddecebb34012e : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12e71af7f0924428b29422975e4a066a : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23a36dd7c0c4c16b4b75d2cb3a1df82 : SUCCESS in 6m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3182f7b93862471da3d4ec35cbae8dc3 : SUCCESS in 10m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3fe5df0c90e4597b6bdbb4a34b2e18e : SUCCESS in 5m 39s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/74ed9a7f045f403c940ae52ce68f2a5a : SUCCESS in 5m 18s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/57ea8667eeba4e28838efa145f55fc30 : FAILURE in 24m 54s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/96cccdeb05e045ab92e0cd28a7bda5af : SUCCESS in 2h 17m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9b368a4c796d46c8ba30fde4bf349f52 : SUCCESS in 1h 53m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/11d25a33cb97452f903f24ea59eeab59 : FAILURE in 46m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/bc04770bfe9d45baa4d8e14531691510 : FAILURE in 46m 02s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3cf1e474957f4aef9c07966f8f76fe8c : FAILURE in 1h 24m 34s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dc2bfe463d2a436dadbfbea9cea6caed : FAILURE in 1h 12m 11s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/4f92b48ac28a43baafcd8c50bfd224c3 : FAILURE in 49m 49s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/f18a8973f8704f1086f07454bd7c7bfb : FAILURE in 47m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3321168bccbe43c9ba772879ab6f5795 : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e71b7175d0b0f9d1ea01db0a08801bd18af1ef58","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-19 08:49:49.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"527b9232129033a2a4be64070492b45b9f0d9e36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 11:36:15.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ee1bf76c4ca425c92c133f83aeec2fb : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11f2d76f0d384061a0b920076e3a7f29 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50789ce70104be8b057851833d9bd57 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bec76644ef8243d69255886c415682e4 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7824af93de9c4bfa8678a457bca0ed5b : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d030b5272ed456ebc33c5a19ee8e4d9 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d6127649b246f29de2dfc61ec99194 : SUCCESS in 5m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec4804194f784eb5908d40c7fbf3fa6d : SUCCESS in 8m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/407bc545b47d4590999b64ce88b229a5 : SUCCESS in 5m 38s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ea677631c895414c84db06a35bffe06b : SUCCESS in 5m 34s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2700d405d11247b989cc3cc792e00976 : SUCCESS in 2h 37m 47s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/eeb7cf6d485b4cedb189d965c05e55fa : SUCCESS in 2h 23m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e09428f27cc64abc8b13109cc3e7bf8b : SUCCESS in 1h 54m 28s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/b4c5e58e904f49bfa1b347781129723f : FAILURE in 40m 34s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e41e291cecbb4b538bb234d0a0f29838 : FAILURE in 56m 09s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/57490e94065d4777ac8fdc78010fb2c8 : FAILURE in 1h 25m 47s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/493a541429ed4220a62d4663406dd2b0 : FAILURE in 58m 29s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/4ab988bb459e4edd97fa882c5bb30a45 : FAILURE in 50m 15s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/22751b62fbe74f1793e5f5e4cbd4c298 : FAILURE in 1h 13m 33s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4f5f0cc37ae4ea09861c0d0c3b2375c : SUCCESS in 47m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1c56c7bb31ee8765d8fbc41a9f947fea67fa240d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 11:36:32.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1c3db6ea925feb5f5274d4fdcd175a90906ec337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:51:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"5258389f3d2012c84f82d286dceac4f4295cbf3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:51:58.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a03219d5fbc244ba8839de3da8be02e1 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9366499664d64192bd3df81065c33c2f : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42768dffbeaf4e958baf2f79de7d28a9 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bc5c15ded7a4524bb1cc9ea5e769ec6 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5483d184e3f344e28edbba65e9a70634 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83f7859fab3402fa6652c54b2fd50e7 : SUCCESS in 5m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36c667b5d8c043e28c1ef52b4ecbe4b6 : SUCCESS in 9m 42s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c3d8b6f01fee4640b4f0e544f48cc034 : SUCCESS in 6m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/105154f359804741aeb90bb28fb53ae3 : SUCCESS in 2h 11m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a9dd7c25a0af40b398adca1bb83282a8 : SUCCESS in 2h 08m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e23c6029dffb4e9da1c1964ae64a27cc : SUCCESS in 1h 50m 04s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1943392c69d34447baedcdaca5416a98 : SUCCESS in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"415968d1d02607c210f7ee96fdd860b1848bd2ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:59:42.000000000","message":"Patch Set 11:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f02811526a74f42bf3ac1e4d2536719 : SUCCESS in 1m 38s\n- publish-trove-guest-image-mysql-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/26b11ce7a7f34f8bb50c5d77e6566671 : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0cdf8b20b664abd26bd25483ead82eaf3640a36d","revisions":{"4d3ee3655dfd24a1f85d1be0711251740be0246e":{"kind":"REWORK","_number":1,"created":"2019-09-16 01:19:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/1"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 01:19:03.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d3ee3655dfd24a1f85d1be0711251740be0246e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d3ee3655dfd24a1f85d1be0711251740be0246e"}]},"branch":"refs/heads/master"},"39a152b06664eafcd8dc8838726dbef44e5eefa8":{"kind":"REWORK","_number":2,"created":"2019-09-16 10:41:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/2"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 08:38:25.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39a152b06664eafcd8dc8838726dbef44e5eefa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39a152b06664eafcd8dc8838726dbef44e5eefa8"}]},"branch":"refs/heads/master"},"60152ae5758481e00a7dcfe1b042c6efa4e55b5a":{"kind":"REWORK","_number":3,"created":"2019-09-16 11:46:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/3"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 11:46:24.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60152ae5758481e00a7dcfe1b042c6efa4e55b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60152ae5758481e00a7dcfe1b042c6efa4e55b5a"}]},"branch":"refs/heads/master"},"5061c094c07baa1d01194ef7c001f01c3c401e06":{"kind":"REWORK","_number":4,"created":"2019-09-16 11:48:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/4"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 11:48:30.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5061c094c07baa1d01194ef7c001f01c3c401e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5061c094c07baa1d01194ef7c001f01c3c401e06"}]},"branch":"refs/heads/master"},"3af9744e291f5b8ac267b4b6834496d7eb84dcda":{"kind":"REWORK","_number":5,"created":"2019-09-16 19:42:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/5"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 19:42:06.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3af9744e291f5b8ac267b4b6834496d7eb84dcda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3af9744e291f5b8ac267b4b6834496d7eb84dcda"}]},"branch":"refs/heads/master"},"05c58dd7587823c5ea1459a45e9e794b57c0896a":{"kind":"REWORK","_number":6,"created":"2019-09-16 19:46:57.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/6"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 19:46:45.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c58dd7587823c5ea1459a45e9e794b57c0896a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c58dd7587823c5ea1459a45e9e794b57c0896a"}]},"branch":"refs/heads/master"},"48d517b3b3c04111890c4e2d926368ad527c8f2d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-16 23:13:42.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/7"}}},"commit":{"parents":[{"commit":"ef72deab749a2c0c6c880d53c8d4845f8a8bed4e","subject":"Remove Designate v1 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef72deab749a2c0c6c880d53c8d4845f8a8bed4e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 23:13:32.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48d517b3b3c04111890c4e2d926368ad527c8f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48d517b3b3c04111890c4e2d926368ad527c8f2d"}]},"branch":"refs/heads/master"},"1a080a634ea605434f3e50fa5b1bba1fd41755db":{"kind":"REWORK","_number":8,"created":"2019-09-17 03:10:52.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/8"}}},"commit":{"parents":[{"commit":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","subject":"Fix Trove CI failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-17 03:10:36.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a080a634ea605434f3e50fa5b1bba1fd41755db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a080a634ea605434f3e50fa5b1bba1fd41755db"}]},"branch":"refs/heads/master"},"8cbabc3d5ce9e2c5db9108d47849e4620af61915":{"kind":"REWORK","_number":9,"created":"2019-09-17 03:16:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/9"}}},"commit":{"parents":[{"commit":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","subject":"Fix Trove CI failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-17 03:15:54.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8cbabc3d5ce9e2c5db9108d47849e4620af61915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8cbabc3d5ce9e2c5db9108d47849e4620af61915"}]},"branch":"refs/heads/master"},"8c214c62166dbb1a5a4690743c8818a6b24f7d2f":{"kind":"REWORK","_number":10,"created":"2019-09-17 10:23:24.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/10"}}},"commit":{"parents":[{"commit":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","subject":"Fix Trove CI failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-17 10:22:27.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c214c62166dbb1a5a4690743c8818a6b24f7d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c214c62166dbb1a5a4690743c8818a6b24f7d2f"}]},"branch":"refs/heads/master"},"0cdf8b20b664abd26bd25483ead82eaf3640a36d":{"kind":"REWORK","_number":11,"created":"2019-09-17 10:23:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/682261/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/682261/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/682261/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/682261/11"}}},"commit":{"parents":[{"commit":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","subject":"Fix Trove CI failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 00:15:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-17 10:23:43.000000000","tz":720},"subject":"Add image build experimental CI job","message":"Add image build experimental CI job\n\nStory: 2006553\nTask: 36637\nChange-Id: I3d909082fb887b24ed25c93e11792031d5053b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0cdf8b20b664abd26bd25483ead82eaf3640a36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0cdf8b20b664abd26bd25483ead82eaf3640a36d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
