)]}'
{"id":"openstack%2Ftrove~682301","triplet_id":"openstack%2Ftrove~master~I29ed7e2d26dde15913088fb71fe4dbcb713751be","project":"openstack/trove","branch":"master","topic":"mysql_support_volume_snapshot","hashtags":[],"change_id":"I29ed7e2d26dde15913088fb71fe4dbcb713751be","subject":"[MYSQL]Support snapshot backup type","status":"ABANDONED","created":"2019-09-16 08:58:05.000000000","updated":"2024-08-19 03:33:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"208bfc8c3f6507259ee8e424f454a9d229d69dd3","_number":682301,"virtual_id_number":682301,"owner":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-16 09:49:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 09:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62359a1d955dbf70f79dd5e25ed10e2b580a7ed9","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-16 08:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc9a95746963159442256c75840d5b1bde9d84a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 09:49:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8bb18f14f3b4b58bc24a69a0a508933 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14805c89eb8c4e96afabe066e0f8f9da : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a60430317ae4f5bacc6111f3edef403 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9486a5fec24b4b60a15fb89f323236fb : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b39dd34d854b02969d1d7197645c36 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5ddf0d6b099487981a10fda8a04b962 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c372ae9a999f4aae9a651a1fab118602 : SUCCESS in 5m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5fb5e2163edf43c7a78f3e5c9562069c : FAILURE in 34m 47s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2d1bbe40683e403caf02e7e181bfbfc6 : FAILURE in 34m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/de9a5defd0714b5ea391b8dad237c9b9 : FAILURE in 33m 19s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d88c2fc6700d4e3486045fb4a085bce4 : FAILURE in 32m 35s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/de8c119fee794c04951493ab454c8a37 : FAILURE in 39m 11s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2744e641d4ea4cce87bf78e179501f31 : FAILURE in 36m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e2a0c0d2d8544216b8da639ad0f1db28 : FAILURE in 35m 29s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/3350048d3d334958b4f9f4f08c6a0ec9 : FAILURE in 31m 38s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/9cb12d4e8ded470f884ffe8e91724250 : FAILURE in 34m 48s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f4eb0415782c429cbd95f2ef5afb4f06 : FAILURE in 31m 32s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/beebe15c009d4048a9daa2df1ac16dd4 : FAILURE in 33m 11s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/9485f2696be741ce8c4df14ff228d5c4 : SUCCESS in 6m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07b5c58a14774085a9950d706d148253 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/67812fa052414994889a0625a38eb1ba : SUCCESS in 8m 40s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4151cad490546728b6c16886e633456 : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"208bfc8c3f6507259ee8e424f454a9d229d69dd3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-19 03:33:44.000000000","message":"Abandoned\n\nno long support. and may be duplicated with: https://review.opendev.org/c/openstack/trove/+/903440","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51e9a222002a0ad919a8d841806150dd9907c9ff","revisions":{"51e9a222002a0ad919a8d841806150dd9907c9ff":{"kind":"REWORK","_number":1,"created":"2019-09-16 08:58:05.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/01/682301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/01/682301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/682301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/682301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/682301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/01/682301/1"}}},"commit":{"parents":[{"commit":"24e6b47bc3822fd42f8fe7b14b8017d36f86899e","subject":"Add volume-snapshot backup type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24e6b47bc3822fd42f8fe7b14b8017d36f86899e"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-16 08:41:04.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-16 08:50:05.000000000","tz":0},"subject":"[MYSQL]Support snapshot backup type","message":"[MYSQL]Support snapshot backup type\n\nMysql datastore support volume snapshot backup type\n\n- prepare_data_volume_snapshot action:\n  1. flush database with read lock\n  2. store configuration files in backup datadir\n  3. sync files\n  4. fsfreeze mysql datadir\n\n- post_data_volume_snapshot action:\n  1. unfsfreeze mysql datadir\n  2. release database readlock\n\nStory: #2005819\nTask: #36641\n\nChange-Id: I29ed7e2d26dde15913088fb71fe4dbcb713751be\nDepends-On: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/51e9a222002a0ad919a8d841806150dd9907c9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/51e9a222002a0ad919a8d841806150dd9907c9ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
