)]}'
{"id":"openstack%2Ftrove~682510","triplet_id":"openstack%2Ftrove~master~I08916ddfba85c0218aeab5bfbad58ce3222521f3","project":"openstack/trove","branch":"master","topic":"fix-ci","hashtags":[],"change_id":"I08916ddfba85c0218aeab5bfbad58ce3222521f3","subject":"Fix Trove CI failure","status":"MERGED","created":"2019-09-16 23:12:54.000000000","updated":"2019-09-17 06:53:02.000000000","submitted":"2019-09-17 06:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682510-1568703082788-b16e9fe7","meta_rev_id":"9986d03de9095aa0f577139ce822d3bff1312835","_number":682510,"virtual_id_number":682510,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-17 06:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 04:14:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-17 04:14:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 06:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33daf0c98487eee63662b71ece91cd2ffa6b4a63","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 23:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8840cc26a48dffb0c6e341d4507d92b47689f6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:26:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e37a805e4cc4745ae7dd0b3f766f78a : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d82b472e3765468899d462d970c63194 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe2f903218a49b8be186b76c9e9489c : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3650fc906de540429a2f0ac5276d8133 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9c004a9ac74ca7ad2745ca90708896 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/779c0e2e125049e896bd34c5787641a4 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94835ca42f45462f82f39c5f82808ced : SUCCESS in 5m 34s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2eae1f546ec14f4eb3a7b0788457c253 : FAILURE in 37m 24s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/747b46c7d39744a6ad2466558a8e313a : FAILURE in 30m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/2ce937292a814f1095ea6600c8840f59 : FAILURE in 49m 53s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2c7882d518f44bc58125f608ab0382c3 : FAILURE in 32m 34s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/737de7b45e0f4b728a4615bc8432a1c4 : FAILURE in 39m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/89d02cb7f6344fcdb777e2b675c4ca52 : FAILURE in 32m 29s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fa22433e7366486f91b91e113bf80fca : FAILURE in 30m 27s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/f281d903dbe44b98b36952bab0c1e9ea : FAILURE in 28m 36s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/099ebf89f008486b8327836d49668a10 : FAILURE in 32m 01s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/411b2be5a63045e994f20c556e620359 : FAILURE in 34m 01s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e2b9c26d5f3e4f06aa551903db1509ae : FAILURE in 29m 31s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/78a806615b67456d8076949d563f7258 : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3495bf1d7864d2ea997d2de4479095e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ada6514c468401a9fbf4f1643e58e15 : SUCCESS in 8m 13s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7494d2a83cd34594a3e72352ee161edf : FAILURE in 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5f2e9128b390766eda7650447f1ff740f25f8fba","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 01:53:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"309a2e53460fbb2bb691ac5ba95063a85502dd91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:12:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bf521ac5049421c858891f06e445b36 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e597eb3d4a49450cb84ce9982e46dcb6 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df37eb6cf78f47b2878dac64370d124b : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10b60ef488fd49ffb12813a13c892375 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12d2f2ede4364bbc9cd27225e558ab49 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e78b3b34f164400b5c9d7ba6f6c79a1 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddea44c0bf645eb8faa0741b36772ed : SUCCESS in 5m 35s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b1c29e338d764ac9b9cdc7e8e1a9cf2c : SUCCESS in 2h 17m 41s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/56360806de664e3f8b45738039785066 : FAILURE in 1h 07m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/04eb667be01348908b44f78e23b6bbcd : FAILURE in 55m 01s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c4bd5244b09e4a048ffa5c0f4a3ac641 : SUCCESS in 1h 41m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/98b6bc2a6c684e9da234b39fca2b408d : SUCCESS in 1h 36m 55s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e641ac1174944b9e803b52901562f542 : FAILURE in 41m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/040844ead4724ce8addd2c3fe5720413 : FAILURE in 40m 55s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/45b50adc0e0b40f4878b9a82f3e27804 : FAILURE in 42m 08s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/affc4f6b07074e6e95f825620d764336 : FAILURE in 42m 23s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/442a3182bfa0414bbefe12202ecb6033 : FAILURE in 44m 48s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/7f6969d35b114c658944d3b4dca41995 : FAILURE in 42m 51s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/106353cc28ea4b058f8062e9e4fc4c50 : SUCCESS in 5m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d27d803a1ff4a3c8c9dfa43ecb3d8e6 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72b7287fca3d467ca6a0f8e22d40bf2c : SUCCESS in 8m 04s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2858aec404b64e43bcc51c4e66d1f9e0 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9c009a710451229307877631fdb7ca7726f51e32","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 04:14:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e870fdf601c8a82d942138e6a878721752c19754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:14:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbf4224e366161af336b15ec95c55e66e08f5b45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:51:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57e2be2bba8d4ecd98864b6d705e3d8e : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a925847a2f7f4527a9a2809f16c94000 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6338c0eb5d8e407aba07d53a8b4081d6 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f79590ba834387a8f0c1cbcc07dd82 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60b37ded7cd54151bc255d9d78bfc276 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fd6eeb49c64ca4b00235c57df157a3 : SUCCESS in 5m 27s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6e3fda5b70b640e48bcb3279ca6b4c44 : SUCCESS in 2h 31m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4a1a8636d2394d08baf22f25459d8479 : SUCCESS in 1h 28m 41s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/876698cb44d343878b13fb7cfae37e5f : SUCCESS in 1h 36m 41s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/580fb1ac21524424b4e04c07c91f4b2a : SUCCESS in 4m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ada60a77ab054b5492c5a2f3d9acd2d2 : SUCCESS in 8m 05s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72b59c07167345ba8e57d29033dd6484 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1e5335d8b193af1aec88f37d9f3c75679d383e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:51:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bb0b64507614f1e591ce3cfa08ce0455d72f340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:53:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747edfc4431840df90d853bb27062229 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","revisions":{"ef72deab749a2c0c6c880d53c8d4845f8a8bed4e":{"kind":"REWORK","_number":1,"created":"2019-09-16 23:12:54.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/10/682510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/682510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/682510/1"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 23:08:00.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 23:08:01.000000000","tz":720},"subject":"Remove Designate v1 support","message":"Remove Designate v1 support\n\npython-designateclient removed v1 support in\nhttps://github.com/openstack/python-designateclient/commit/093d8d7170cbf6ef8c7a7c0ff2a4dcd7ecd6361b\n\nChange-Id: I08916ddfba85c0218aeab5bfbad58ce3222521f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef72deab749a2c0c6c880d53c8d4845f8a8bed4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef72deab749a2c0c6c880d53c8d4845f8a8bed4e"}]},"branch":"refs/heads/master"},"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1":{"kind":"REWORK","_number":2,"created":"2019-09-17 01:53:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/10/682510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/682510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/682510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/682510/2"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-16 23:08:00.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-17 01:52:21.000000000","tz":720},"subject":"Fix Trove CI failure","message":"Fix Trove CI failure\n\n- Remove Designate v1 support. python-designateclient removed v1 support\n  in\n  https://github.com/openstack/python-designateclient/commit/093d8d7170cbf6ef8c7a7c0ff2a4dcd7ecd6361b\n- Fix get port IP address error in devstack script.\n\nChange-Id: I08916ddfba85c0218aeab5bfbad58ce3222521f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
