)]}'
{"id":"openstack%2Ftrove~684567","triplet_id":"openstack%2Ftrove~stable%2Ftrain~If66b5d738173181054f934d35e32f5c5b69a6c6c","project":"openstack/trove","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"If66b5d738173181054f934d35e32f5c5b69a6c6c","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-09-25 08:41:54.000000000","updated":"2019-09-26 19:40:37.000000000","submitted":"2019-09-26 19:39:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684567-1569526748798-efb49f7a","meta_rev_id":"32a1e6ae44d518244ff224d5c8e4421f934d373c","_number":684567,"virtual_id_number":684567,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 19:39:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-09-26 19:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-09-26 19:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 19:20:46.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-09-26 19:39:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ef662586f42c7c5a14e0bfc3c4e38fdcb0c5bfe","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-25 08:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46034e6ba217420380bafbfdfab5017879f590cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:25:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f49bbd63e8d455a8c542f0e75763318 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/446dcc86aae4467dbf9742435991116b : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e3a55173fe48418edeb15e64b3d209 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/908189e36f0447fd97a638bd7b6ca3ef : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/475550f7f5eb4ce8b258610c6283927a : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aee79e83ac4641bba5424dba04ad3a8d : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25239fff9765477885867dfa6898f460 : SUCCESS in 6m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c754b8fd72424622b6a158a515de3baa : SUCCESS in 8m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6848da337fc74ee1a9124cebaf67a07b : SUCCESS in 6m 32s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ab0afc2d47b94a6d99b4715b2a794684 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"086e43605a07ea236589ff4be4fd7fe4e131f9f9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-26 02:13:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7b1fdda417a190e875bfed5f2f3203f0bf39136","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-26 19:20:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74cfc582749521581d72bbfd26e277cfdbafc76a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:20:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc48764e7b85e24dc6285ee8e0515e2e7e061657","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:39:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f5e01b44222ac43de0ff41624240a17dc159d2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:39:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b42ed724b4384a179485aa51aa3bf237 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f5033afddf43c98f480bdad090a048 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/928a2de40bce4959ab52a331c8c135c0 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0860d2cff14318b6a55156207e8ec9 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9385109c13fe4a39afb3bc974741eb1d : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1b073eea0444909595e4b89dc979bf : SUCCESS in 6m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5c76244e4ce40cc9700bb63d0ed9a34 : SUCCESS in 10m 39s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/76bb24b7bfbb48bc8837d4d0482066db : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"821dfe6cd5a63a5d30aad4f05bc2b421ad316b47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:40:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbdf82ad96d44b5bd3746e4f06fd983 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd6ca3e1c7db551efe0138aa0862d6841531687b","revisions":{"cd6ca3e1c7db551efe0138aa0862d6841531687b":{"kind":"REWORK","_number":1,"created":"2019-09-25 08:41:54.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/67/684567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/684567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/684567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/684567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/684567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/684567/1"}}},"commit":{"parents":[{"commit":"59be66ceaacba7be038ff17817fa6bd8e0aed4ec","subject":"Change imag build job to the periodic pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59be66ceaacba7be038ff17817fa6bd8e0aed4ec"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-25 08:41:51.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-25 08:41:51.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: If66b5d738173181054f934d35e32f5c5b69a6c6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd6ca3e1c7db551efe0138aa0862d6841531687b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd6ca3e1c7db551efe0138aa0862d6841531687b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
