)]}'
{"id":"openstack%2Ftrove~684881","triplet_id":"openstack%2Ftrove~master~I7874d498eb80c0b5cda802acb76e5fe399770e6a","project":"openstack/trove","branch":"master","topic":"branch-image","hashtags":[],"change_id":"I7874d498eb80c0b5cda802acb76e5fe399770e6a","subject":"Add branch param for image building","status":"MERGED","created":"2019-09-26 02:32:52.000000000","updated":"2019-09-26 12:30:06.000000000","submitted":"2019-09-26 12:26:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684881-1569500761294-94a46b85","meta_rev_id":"c87234a382aac6f314efb03a86dd592cbed425a1","_number":684881,"virtual_id_number":684881,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 12:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-09-26 09:47:24.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-09-26 09:47:24.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 12:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"561dbf93066bca586be34c8f353711ee8e707777","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-26 02:32:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01276459ac653ee9c7c000c081a792ea5a7046d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 05:21:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6a70e54677d42daa71e60a6e8df0491 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fe010d0ad13427d85f04a5adc813a41 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ba1fa5c49a43f7a79f62f701678b25 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56b7903b8cdb4a79a74a60354a9884f9 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ab4be0d2fa4ff4a677a948f4078750 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3597a25466cd4cf092dd0a156e115d24 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0381109828584a0d992cbfc97bbc0807 : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b26f69eefdc54dfba41979fafd83bec6 : SUCCESS in 8m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/646ead5e4c2e4409801b27fc2f2a0ea8 : SUCCESS in 5m 45s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ca4e5511505f453caa1abe5b1cd31532 : SUCCESS in 4m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/329e0cf48199472aa67bf4eb876b4db0 : SUCCESS in 2h 43m 09s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/32d3cdd4e729422b93f25959fda68f47 : FAILURE in 1h 10m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1f6e5dba39bb42f38068f1e0f7f14ca5 : SUCCESS in 1h 47m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/23ab55ab24fd4260b93c329df889d87d : FAILURE in 47m 24s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7afce6dc4e944fe39733c991ec141940 : FAILURE in 43m 54s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c6744dde71bc47f5a932db9c7e450232 : FAILURE in 1h 15m 52s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/27146669bb1a489594b93dd4c444720d : FAILURE in 55m 17s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9f39e1731f974b38bdf3fa6524d83e62 : FAILURE in 44m 34s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e222cd64198a4a2e94547fb5b2280385 : FAILURE in 40m 32s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2da52cccba244c5e968d959d8106e39e : SUCCESS in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4e7f3717523b07fa93556f2aef7e81cdd696dda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-26 06:25:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"faef57b23dab3e39ad8162f4fc7be298aae2ecac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:08:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0605823457f24ad6a8e1ef282587f248 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/133450d470ce4c13befa9a763eac4010 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77177c91ecfb40b7946352929d5dc20a : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26e425e6da054041a917af9be9c2c0e1 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aadd9eb4e044ef2957c9366c3ac6cc1 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7496d381e3854ec6b44cad7ca29521cb : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeda8f77e2a94728975f475abf622471 : SUCCESS in 16m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2eaaba873b17450f93f70ae017c751ff : SUCCESS in 16m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/44149d8bfbd94ea89a1b0e414d748fb6 : SUCCESS in 9m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/853bfe1c16ce4c73ad1a5e3717354eec : SUCCESS in 6m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4dc5be55a6894fc9b7787c680f964917 : SUCCESS in 2h 22m 24s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0d413c94cc0f4677b2d0d983ae39c4df : SUCCESS in 1h 46m 51s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7b0f1b55be9c4f7b9beb372e3f3ec5b3 : SUCCESS in 1h 43m 57s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7e3ab634ef2e49e99a1e80252766976f : FAILURE in 40m 48s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/be80ab618f984d21bb5ee4a81cf35b97 : FAILURE in 45m 30s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fbaded154d374acfb363f5736bfc3efc : FAILURE in 1h 17m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/142c3df61d2b4ecd93143ac5b4d3ac73 : FAILURE in 57m 00s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/18659a5156b744ef95b5cc4355d6c095 : FAILURE in 1h 04m 48s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/eedf9570461742e6977b571938a13f0f : FAILURE in 50m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59fba5b87a2544b29bfb70e89cbe9114 : SUCCESS in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cf9c6dd65713c6497236974723d8fb7bf9a62b8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-26 09:47:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"021871b73ab40ade5520794e0e98b3e9a9fbc5b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:47:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"062c772666c7eab9d1072c2d5f57d5517223370b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:26:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88421c4310654cf5af596152d49ca336 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a82e0e28cf4677b517593c87f82638 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2414996687354d7ab01ad898984529d5 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02d7e0c0a1064d57a47c2bb9d77c2dab : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b8028c147144ce4bc40e084095a1238 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3708976784425c9b2920b734659f2b : SUCCESS in 5m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2522ec948c7e4743961da7afefb6d662 : SUCCESS in 8m 21s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/56e2713aa7b74bb8b38f350dbc9fe99b : SUCCESS in 4m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/64932a3f128d45deaa7428e90193d970 : SUCCESS in 2h 32m 40s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/9e1cb756dcc14bb79df8d889dd88e60d : SUCCESS in 1h 44m 20s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/739a06565074447f9cc1abd12746c716 : SUCCESS in 1h 44m 46s","accounts_in_message":[],"_revision_number":1},{"id":"923aeab35a99632df4fa8c3d41a2645f62185c47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:26:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55f561e17c948bffa1ca046c0b4a87bfe3b1546e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:30:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6263d7158baf467b8a25b912ae18d0c6 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"062871b7edd03846f2f9d40e085d38e385f262ff","revisions":{"062871b7edd03846f2f9d40e085d38e385f262ff":{"kind":"REWORK","_number":1,"created":"2019-09-26 02:32:52.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/81/684881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/684881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/684881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/684881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/684881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/684881/1"}}},"commit":{"parents":[{"commit":"f729b9e05d0efa54bc97853a283f0a4db0b5d6d6","subject":"Merge \"Add release note for public images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f729b9e05d0efa54bc97853a283f0a4db0b5d6d6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-26 02:32:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-26 02:32:39.000000000","tz":720},"subject":"Add branch param for image building","message":"Add branch param for image building\n\nChange-Id: I7874d498eb80c0b5cda802acb76e5fe399770e6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/062871b7edd03846f2f9d40e085d38e385f262ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/062871b7edd03846f2f9d40e085d38e385f262ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
