)]}'
{"id":"openstack%2Ftrove~686468","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I5859f4c9911fc2c129c8f23611c6607044fbc145","project":"openstack/trove","branch":"stable/train","topic":"remote-swift-client-stable/train","hashtags":[],"change_id":"I5859f4c9911fc2c129c8f23611c6607044fbc145","subject":"Use correct Swift credential for instance backup","status":"MERGED","created":"2019-10-03 19:22:45.000000000","updated":"2019-10-04 10:21:22.000000000","submitted":"2019-10-04 10:20:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686468-1570184418081-2f73a324","meta_rev_id":"ef666eb208df55309fef87f698faa9f98bcc20fe","_number":686468,"virtual_id_number":686468,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 10:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-10-04 10:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-10-04 10:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 10:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"907a4d75ca61894f242f397523e0204223fa0b50","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-03 19:22:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6090cd430bec6862edfaf49c8ecf1d0467e12318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:35:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20d25623c74744eba9f9b25f7b189de1 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/921d923600694e8783f8cacd6096d664 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7701493fece547f99d18002f7ebc7d69 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89324b8975d1408da6af78841b9eea2e : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b9d9200c93438cbc908df5303c8144 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e049497fb7c842b49e3a1c77a5f83cc3 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03b69caa60e49dbbf73b6e24b62acb3 : SUCCESS in 5m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae495fa814f942cea550cae1d5208848 : SUCCESS in 8m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7a3e4d308184f9ea98128f98b45e065 : SUCCESS in 5m 33s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0d160301986147b48a2c92c5d0f5b4f2 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e67e77056f1f5c58139b8f90a335592be8829b68","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-04 06:45:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8534b02b07c1fe781902032d8743132d13c37f29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 07:03:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd36331ce66447cb9b6bc2e8464385db : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1d45b7a4d54465a85da3c422769b51 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532c653c97ca4c7c82fa7b439b342fcc : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d51846ebddf94f27b40ea1a095e305f1 : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b7bc5db821e4ddfa385eb81221788be : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1ecac283f0d485a8011cf793f3f18b5 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a380e14024244bdb2f5b4777796b495 : SUCCESS in 6m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/923874c622fe44b5838fbe02dd4a2740 : SUCCESS in 9m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c489583c94941f19e18cf3a0d822525 : SUCCESS in 9m 40s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/35dbc64f43ec453cac7bef81b87fd4e6 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cd9b6d3df925c618303e67f4f6e8a96532cf27cd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-04 10:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d43ccc14e113a734a9eec28acfc028666ce1a3d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:03:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdab5f4930a4437d6fe8c0f509469a7362fd702a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:20:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00445920f92441fbbd01c0d629f8297b : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5625d13513b04f698f43e2775869c1bb : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5185cebf9adf4549b3696c605f2870e5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f36b0bc7b7ee40b0a6013a0826d0139a : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4536c657401473eb90f01fe4d8493b9 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56547b0a6fbe45eb9113b42a602351e7 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21812a25f2064960bb4b10b7ce58c1be : SUCCESS in 8m 35s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ee60b51b1aa9401a8beb319ed4c13fa6 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"32eb6397b2df9967f0abbd4c838ab00d71b5847f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:20:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aeb956dc154c395a14f48f239199cfa834f29669","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:21:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c997ba64204777a94b119cba90a8d1 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e45b6a63490649ac0b3fbce27e7c79d547814840","revisions":{"e45b6a63490649ac0b3fbce27e7c79d547814840":{"kind":"REWORK","_number":1,"created":"2019-10-03 19:22:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/686468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/686468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/686468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/686468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/686468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/686468/1"}}},"commit":{"parents":[{"commit":"e8a6a60f24abc76b5b040fe3dee916865b9623d4","subject":"Fix the stable branch image build job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8a6a60f24abc76b5b040fe3dee916865b9623d4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-01 10:41:52.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 19:22:45.000000000","tz":0},"subject":"Use correct Swift credential for instance backup","message":"Use correct Swift credential for instance backup\n\nDevStack is deploying Trove in service tenant model, for swift backed\ninstance backup, the swift container should be transparent to the end\nusers.\n\nStory: 2006647\nTask: 36883\n\nChange-Id: I5859f4c9911fc2c129c8f23611c6607044fbc145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e45b6a63490649ac0b3fbce27e7c79d547814840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e45b6a63490649ac0b3fbce27e7c79d547814840"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
