)]}'
{"id":"openstack%2Ftrove~686955","triplet_id":"openstack%2Ftrove~master~If0991e0cef40ca78752fcd509d8438e90c9557bc","project":"openstack/trove","branch":"master","topic":"fix-port-left","hashtags":[],"change_id":"If0991e0cef40ca78752fcd509d8438e90c9557bc","subject":"Remove all the resources when the Nova VM creation failed","status":"MERGED","created":"2019-10-07 02:16:33.000000000","updated":"2019-10-09 09:21:23.000000000","submitted":"2019-10-09 04:16:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686955-1570594586723-d762e54f","meta_rev_id":"1f25d20bb17abbf937e452251ed6f4c676b1e423","_number":686955,"virtual_id_number":686955,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 04:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-09 01:39:18.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-09 01:39:18.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 04:16:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc004490a50d1f76645cd01402101de7f2dca7e5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-07 02:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3351bd9c1e520bdff8bfd85ba462a59520f3d471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 04:54:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95b5667d8c584d9abc601bccb60afb06 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caea3f9663fe407d845dd88162d31fd7 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883c65d5dc4c408ca6f0a963c28ea031 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ccf6e31b8dd44fbaa4b910b6bf1aa93 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab6c84d2f4c43f0ba7b2f1754fd3b96 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6e4f5cf6dd8454093fa7d45d66e1e12 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dbfb793730a43d285c130aeaaf56d7b : SUCCESS in 4m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ac455fbe19f47b58d65bec82bb51c0f : FAILURE in 8m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0dd4b5719cd746d1989025141d038669 : SUCCESS in 4m 58s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/61775494c52b4a6da3a8417efe4e32bb : SUCCESS in 4m 50s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5b830098148b4dd2a0a6dabe924acaf5 : SUCCESS in 2h 36m 17s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/27e730ad88c3473e88aae2271b68cf82 : SUCCESS in 1h 48m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/deba1dd941d6460e96a0530a047645be : SUCCESS in 1h 36m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ba22418b23c84cd8a3082e6af7b1f78c : FAILURE in 41m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/34bb2b35e2fc4178b4bfb42211cc3538 : FAILURE in 38m 28s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f838adda71a944efb068195648768b1e : FAILURE in 1h 16m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ffd30a833f75494988476f33f97cc8b7 : FAILURE in 54m 28s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/b9db21c4b0dc4de3ac8d81f68ebf17be : FAILURE in 42m 50s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/cf4c029c9dee4704b65ab6c089dce697 : FAILURE in 40m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac867ab167dc4330a5964d55bfc22762 : SUCCESS in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c2eaa9edd393e8e5cba3d7daea1bde2054ddb92","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-07 09:23:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e998db6e07818c5b3f6874bc389ab4313a038de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 12:31:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0150e2d8863545ee9ea10184fd1aa46b : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d680ab5818804a7581b124544f9ab8cf : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadd022abe2a4d9997c939d48533009a : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3548867ec2d342dca6d632feb15868f4 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c82f5a5cba8e4c048fc0106654712d3c : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b23fe92c8fb84a81b1e803a5f018bb9d : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6584c2950d54811bea98e1faa26e179 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fca7b1dfda8a4c9c8188f5374831d4c4 : SUCCESS in 9m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f64b46335de1482a8578f215701ba56a : SUCCESS in 6m 12s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/42d6c935b55443b1985e3bf84eda71e2 : SUCCESS in 5m 51s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/44a5ab3513194030a7b68de7f1108579 : SUCCESS in 2h 30m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/296ebc0d3fc140d2b45a408e4166e2d1 : TIMED_OUT in 3h 04m 55s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f98268dfb778401aabdff9589877d442 : SUCCESS in 1h 25m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/60667cc030dd48e7ac3c7973e9024824 : FAILURE in 35m 10s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/cef6d2687ff340699fa7ea506f409df9 : FAILURE in 38m 28s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bac3046828fa4a6a960a90e1800b7b28 : FAILURE in 1h 10m 41s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/555f719ef3ea467eb0f0e878693ca070 : FAILURE in 56m 40s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/955207cde1e94ada926a91d5694cefaf : FAILURE in 42m 47s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/78de3eaaa63043bbbdf2f750a10aaa73 : FAILURE in 42m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75faf08186dd4f078ef55d85f44bf069 : SUCCESS in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87d1795b4d0dc808d989c94962d287b355579cf8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-07 19:02:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3e0fe8eeb8c6552548c81acbda43da74db9e25e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:42:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce08429d4d74903ab97baaf5543b1c9 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea7069dc24804aac860c2657f4831aa6 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902e78a75b8046c7be8b6290dbb84632 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fb3c1038a72471a8cee9e21bb158be9 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f895f9f6b2a484fb89216a5862400d6 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1425a4e4cba544569e8f5b7159877551 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb775d686f134d2c905b23fe10f188ee : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c533008ddf4b4964aa1ef67e1486fe40 : SUCCESS in 9m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67ff2218b3414be596d485a2d77637eb : SUCCESS in 6m 00s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/7bc9b90ff3ba4202bfff18d3423734b8 : SUCCESS in 5m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2d765d70e3c847d5b077a94ce5597936 : SUCCESS in 2h 40m 22s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/09c241df72ef400596191ce3ad5fa535 : SUCCESS in 1h 48m 27s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/00c3602c5eb94c16a090d541614a82db : SUCCESS in 2h 00m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d4c61a0359504c2f91553c27032fcb9f : FAILURE in 41m 03s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ca31edfac9524afca788c22720ecccd9 : FAILURE in 45m 13s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/9e4f5d5969334b4a8bf4f555dc6c1167 : FAILURE in 1h 16m 18s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6ee1752aef364c14aad6cdb5cbd7ea5d : FAILURE in 1h 00m 01s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/4a058502778f450694bd064bd2e9216b : FAILURE in 50m 45s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/bf4f83e751b245418a9a7bac1d439c04 : FAILURE in 46m 04s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69a5fef5228845de86938920c8348b93 : SUCCESS in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df7a7c4b077ee09a55bc11ada26e426361052d0e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-08 03:01:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f075deeaa2b5dca4aa4695103200f98e5c023c6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 03:01:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d831fb48c33a708372f8eaff25e7e9d34b89c60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 05:41:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375a1f3046cc420f906513e27c84ca33 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20e8e5d87e34f2bba64e193398d373c : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beca9c619a4c4d1b9370dc3517c787cf : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90d926d4c14a411b97267586fb97a97e : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d581bf14a5bb483db5020f33b62a2e0c : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8b797e2e8848b58515ea644d63ec75 : SUCCESS in 4m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48bcf44ec704436e8ee11a740126be9b : SUCCESS in 7m 41s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/55ad9c0dffab4659bba45175bb361af8 : SUCCESS in 4m 44s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8b6b612507f845168251af867ce0d396 : SUCCESS in 2h 24m 59s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3f0dec0ab2bd4e0fad664c453abaa784 : FAILURE in 1h 31m 16s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2e88f955c44d4314b1bdf807bdec4196 : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2b493759a7092192b72e1cde5e3e292b0bfe745e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-08 08:34:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9911ee6d02b04b7e83b7c8ea15e6ab7b53abe7b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 11:12:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5474c4918974fe188344b97e49a1e42 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37883e89094c4bc6ba28e00d4ce812fc : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5f9b33125642f1bfca0690946b9155 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d01db5251624cc182ecde1fca86c98a : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/775d49509daf46fb82cc75a0460af4bb : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/998e9a923ec04996b793226ef9245bda : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab04770feb45400d826371ec8c0158fe : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea06d4cdbedd47a8b97e74ff1cfda437 : SUCCESS in 8m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73dc838b5c5947bbb75e4f9e4dedc316 : SUCCESS in 5m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2733c6ccf1f5435d82fb0c953ba73b55 : SUCCESS in 5m 22s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/979c195f3f2e4d958fb56ff47f38dd17 : SUCCESS in 2h 35m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/acdb6ddd176a4309b0919d9860c5a555 : SUCCESS in 1h 41m 55s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b1e4ec7bd5494be8af795b0a8677a8e4 : SUCCESS in 1h 33m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7f41f6966b14447baa7aeab2c0d422d4 : FAILURE in 40m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ce6b0879722e4fe5b768d6ee10b77b8f : FAILURE in 39m 51s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/826ead2aec3e4c0a8ba7f1996fa74205 : FAILURE in 1h 22m 53s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/87a75f16324b4b53b1d72bb3aa7aeb10 : FAILURE in 56m 36s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0878408983784ee387ac361b68a8f262 : FAILURE in 44m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/b6fa1b3822764d818d9de1c394559921 : FAILURE in 54m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f3d35fbff0e4753a3b72e33cbeb9926 : SUCCESS in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3edf58289951c767b4de378250baf7e99ce543e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 11:12:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"568c618c0850b2e5bf8e63866de99281537c34eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:07:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/034800cb020d4036948989dd9a5e505e : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a67d2bc59de441392b8c28774478f7b : FAILURE in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffce34efabf049768ae8d6c2b4834e2a : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a83652d18b840c391d3be35b09e4366 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21d176e30588457988c03afe224673af : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb26c14121b49639ee85c8f41ba6b53 : SUCCESS in 5m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e5664379aa441ca84e73986a6bb8ceb : SUCCESS in 8m 29s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/a6caa19d5ca3415db49ec202bb1df4f8 : SUCCESS in 5m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f516754fda35426fb9863bd52034656d : SUCCESS in 2h 52m 29s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f4381f47ff4242eb9cccf001dd80b02f : SUCCESS in 1h 36m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a024076015ef471f96d4fc6049b0d462 : FAILURE in 1h 41m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0b21cedec916911e439b79a5f7f86b67370cb9ca","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-08 19:12:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6f15a4fd782fa98f90510b1698e1b14ac4eeebd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 21:36:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8b69b60433b4e399c9d6304d11e9db6 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb1b8714841e4ed6ae84f79b46f46a52 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3385e6e7537b445d99ac2781e25f0c5d : FAILURE in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efb496f2867a4ace99b1460b61ee65ce : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6829db10e2384c1c9a86d3abef29ff44 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b4b059ab4a04180b30e22f932784bce : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bde9f2479b943a68f4730d430a8d592 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d31de394b87a4e8fb344ab5d42c82cb6 : SUCCESS in 8m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b02a721b4f684850ba1104726b846ff9 : SUCCESS in 6m 22s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c242c77cc6ad44d8b7886781fb3e520b : SUCCESS in 6m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/55f60bfe6a2841d9ba1b1ec113773c7f : SUCCESS in 2h 17m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6e090d59b3ce4e6f8d34c74178f1b703 : SUCCESS in 1h 54m 24s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a7d7ef925c1042479c646c11ed39d04a : SUCCESS in 1h 52m 14s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/905bd7c975f64ba0934809ab821026f7 : FAILURE in 49m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c67b7c53ae9241dfb12f581f12d9b6c9 : FAILURE in 45m 02s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/673af1f5d9a94ec98240a0841e7d194b : FAILURE in 1h 16m 19s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/413611d3295f41c4b3b2047ed2df8672 : FAILURE in 58m 11s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/04f025fea11b4bea9107bdfc9e7d34a2 : FAILURE in 44m 06s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/d8c058f3dc3340abb5b76397f71fbb58 : FAILURE in 45m 22s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bdd896ae0ba4a09930991a6bc2211ef : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83e6a3d815c445cd2685a1d41be72e4ead67a4a7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-08 23:01:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6aad0391194caeb8b5ee5bcd71bb13969b49b634","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-08 23:04:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7db9ea31b17e2861831808cfc4abe99652eb169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:37:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1b61c0ea5cf4bf3ad3b5467c8a20dd5 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2fbde8f37d94494ad7bf286cb0d04b8 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca232f2845b34978ad0056ae7cf62979 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1127786290b846f580d82b93768c99f9 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae93835920c41f9be344296912f5cf4 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c6c490c4cab41e681904f55e6b3937a : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbbedbd4ba64af882ee987c96d9b6e0 : SUCCESS in 5m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/513ba2e57f7e45fba48b86c7effa2d66 : SUCCESS in 7m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c484e8496aba4d248d2f01fd0a5ad8b6 : SUCCESS in 5m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/fa1a1d22e55a4ed7a4cb29a5b97e2bea : SUCCESS in 5m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e200324381234db9aae91d99aa6cdead : SUCCESS in 2h 27m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e7a21e1f143d4b089cc58bfdf8c3d60e : SUCCESS in 1h 47m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c480c97394634a348c19e09582269c30 : SUCCESS in 1h 37m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/651b6b22b1034721ad1585a82a2994e7 : FAILURE in 33m 08s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/961abaefd2c949d5bc94f05a6c134ac8 : FAILURE in 47m 26s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e94a71a9b250498d82f0114d5f5d4161 : FAILURE in 1h 16m 06s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/92d22c5c10374f85a242e4fd19abc774 : FAILURE in 54m 03s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/6ed280ce1bbf44d8a316cfea0d12bb06 : FAILURE in 42m 25s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/d58dbe8e0cd84f3eb5008bece912fb11 : FAILURE in 46m 04s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88e2c7bf037d41efad0e43f72e33ef1a : SUCCESS in 36m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54bc7c8151f3e5b94d468c05cd2639904bbc68cd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-09 01:39:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9646ee462f11960ca58e9e24124f512321dcaf6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:39:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3247d7b93c243aceb4d66aa868206664f2d054cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 04:16:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2268eba96f4843e399a8c1bdfdf584a1 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b09c80289eb4991a3456ee57c29d0b6 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afa2941ac11e4cd2b14085efa2393953 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22935b8489e84eb4aa2cfc32764b1887 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb18594327f74e78b163e86ff82dc64e : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d346ab7d734a6d979a4a3e020b36ee : SUCCESS in 4m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b9bd1ac62bfa4ca18533eb0035dea959 : SUCCESS in 8m 22s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/aa398505bbc24a75aa088d7e429f1f69 : SUCCESS in 4m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c93b7ec4ef654078bc1b6fd21c2000b1 : SUCCESS in 2h 31m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d2fac8a7b77b42c3bac29fd2a807f086 : SUCCESS in 1h 46m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1f2df798626945889b0ac139bae84056 : SUCCESS in 1h 54m 45s","accounts_in_message":[],"_revision_number":3},{"id":"43d77a1e161a1f7566443651921afcacf975b41e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 04:16:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"41f3b555c8e44660b97d2a0bc4028b099f75d949","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 04:17:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ef88ef151d438282933da7f750c186 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3},{"id":"dff53f971073fa0fc740805ee0b70be1c78326d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-09 09:21:23.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6bfa510194e3cf30c8c03446d435edd87c1b4d23","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"650794eaf9f328bd1b2fc096992e6190a9d79c87","revisions":{"517549b0005d6a337465706006a638a0be9a58d5":{"kind":"REWORK","_number":1,"created":"2019-10-07 02:16:33.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/55/686955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/686955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/686955/1"}}},"commit":{"parents":[{"commit":"77c409c0cb5b63f5d3022cefd2e36c171c6968f9","subject":"Merge \"Use correct Swift credential for instance backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c409c0cb5b63f5d3022cefd2e36c171c6968f9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 03:48:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-07 02:15:16.000000000","tz":780},"subject":"Remove all the resources when the Nova VM creation failed","message":"Remove all the resources when the Nova VM creation failed\n\nBackport candidate for stable/train\n\nStory: 2006664\nTask:  36926\n\nChange-Id: If0991e0cef40ca78752fcd509d8438e90c9557bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/517549b0005d6a337465706006a638a0be9a58d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/517549b0005d6a337465706006a638a0be9a58d5"}]},"branch":"refs/heads/master"},"a5772f83a0f0c53e170cf13af7d4b87a4a377ea0":{"kind":"REWORK","_number":2,"created":"2019-10-07 09:23:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/55/686955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/686955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/686955/2"}}},"commit":{"parents":[{"commit":"77c409c0cb5b63f5d3022cefd2e36c171c6968f9","subject":"Merge \"Use correct Swift credential for instance backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c409c0cb5b63f5d3022cefd2e36c171c6968f9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 03:48:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-07 09:23:15.000000000","tz":780},"subject":"Remove all the resources when the Nova VM creation failed","message":"Remove all the resources when the Nova VM creation failed\n\nBackport candidate for stable/train\n\nStory: 2006664\nTask:  36926\n\nChange-Id: If0991e0cef40ca78752fcd509d8438e90c9557bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5772f83a0f0c53e170cf13af7d4b87a4a377ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5772f83a0f0c53e170cf13af7d4b87a4a377ea0"}]},"branch":"refs/heads/master"},"650794eaf9f328bd1b2fc096992e6190a9d79c87":{"kind":"REWORK","_number":3,"created":"2019-10-08 23:04:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/55/686955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/686955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/686955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/686955/3"}}},"commit":{"parents":[{"commit":"77c409c0cb5b63f5d3022cefd2e36c171c6968f9","subject":"Merge \"Use correct Swift credential for instance backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c409c0cb5b63f5d3022cefd2e36c171c6968f9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 03:48:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-08 23:03:59.000000000","tz":780},"subject":"Remove all the resources when the Nova VM creation failed","message":"Remove all the resources when the Nova VM creation failed\n\nBackport candidate for stable/train\n\nStory: 2006664\nTask:  36926\n\nChange-Id: If0991e0cef40ca78752fcd509d8438e90c9557bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/650794eaf9f328bd1b2fc096992e6190a9d79c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/650794eaf9f328bd1b2fc096992e6190a9d79c87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
