)]}'
{"id":"openstack%2Ftrove~687507","triplet_id":"openstack%2Ftrove~stable%2Ftrain~If0991e0cef40ca78752fcd509d8438e90c9557bc","project":"openstack/trove","branch":"stable/train","topic":"fix-port-left-stable/train","hashtags":[],"change_id":"If0991e0cef40ca78752fcd509d8438e90c9557bc","subject":"Remove all the resources when the Nova VM creation failed","status":"MERGED","created":"2019-10-09 09:21:23.000000000","updated":"2019-10-10 13:10:22.000000000","submitted":"2019-10-10 13:08:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687507-1570712917608-e8c316f0","meta_rev_id":"0b640295bfd95c0e086b4b6309a573a5ccbf5d41","_number":687507,"virtual_id_number":687507,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 13:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-10-09 21:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-10-09 21:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 13:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f859f9eeaed0b8dd8c17cd49199dbc3cfeafa8db","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-09 09:21:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b36aa615ae458f4ef8c443b388dfe89957265e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:52:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec16030acfd54f359d66e58370874783 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beece7db634d4dd3bf537ee015b08983 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865b845b5a4b48d0bd872268d69664be : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c09b536f7f640168713a7ee071f13e7 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/785f2fddb2664e858050ff185ab504fa : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ac525294cc440fbaea82ac921130eae : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384c1ccc769a44b49abc97ff84881c02 : SUCCESS in 5m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97390294f6df448d913e7d77c552c19f : SUCCESS in 8m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c44e558aa4941b19fcfeaf777cdeab7 : SUCCESS in 6m 06s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/985c10ac815349d7914dfd2dfff615e1 : SUCCESS in 5m 23s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c8640f50dd4c431e944a15ee3cb48e74 : SUCCESS in 2h 30m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/15d3c46afa224265978e6f3f7d967abf : SUCCESS in 1h 59m 44s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f50f733d11a945b5ba3cda66346706fc : SUCCESS in 2h 22m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/863c298ed13e42d38f8844568b18e22b : FAILURE in 45m 32s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/be46b3ad0cb04007bf021def2307bef4 : FAILURE in 47m 38s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/8c632ba2e6e24752961642461d0ac1f7 : FAILURE in 1h 13m 50s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7102039d5b2040fe969c2198d65b6d68 : FAILURE in 1h 02m 02s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c19051e68969499ea492990cb1781a4c : FAILURE in 56m 33s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e1d23a65ed2747b0a77896e0606cb106 : FAILURE in 50m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fac393c4d69442bb04cbef1d3bf195a : SUCCESS in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a39a23fe497a002df51915c157bfc0069050c4e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-09 21:06:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6531ebf0075ed535fabe533db867db2d15dc3a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 21:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd2eed76dae8a9820c329b7a42fa976c870bb598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 00:17:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2e8a0f845e64c5995bd8e8665997b70 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e9195fcaab4294b76693ec3a08ae5e : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e92154e2ff13467d84b1368af3b014ba : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f0d1e23b9c4f1d83f97ca95288f6ae : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e88704da4e9470ebce48743d83e717f : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b2e929db664839ac7a341608655af7 : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b214477af57c4d348e2eadc2fd4c8dba : SUCCESS in 9m 14s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2cfc7e250fbb4c84a3c5ecb6a90532aa : SUCCESS in 6m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2786ca1126664065bff7375e440a733d : SUCCESS in 2h 23m 36s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b6de7ae195f9468fa0c56ce4119e3332 : TIMED_OUT in 3h 03m 36s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2e4506b0cb034cc1b3ba7361f1587504 : SUCCESS in 2h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a3d14e0906674f7ff5e196808584c003c3705e1e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-10 02:33:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6cbd2acad6f32943a679157c3ead85c4561dc34d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 04:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c51b34efee4b61a1543c261de5e19f : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1e3a1a61d4e4d85b9eecaa196d79a8d : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ab0fded27a41c5a07da70dd771ea34 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d47b0bde28742d2b192dc85b7c52846 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f5ac1c19196405180477965a41a8965 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7321744cec614f25b6f3328e20273035 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cc8ec10772456583bf73d7aeb588b2 : SUCCESS in 6m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b94628d460644ff7b230f8f0d751f95c : SUCCESS in 9m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c452561ebb54df38679d334471a3557 : SUCCESS in 6m 26s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ee945ea2a65446eea5f916e4f2d9b5de : SUCCESS in 5m 25s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1fcfb6c719c94c9987cea223c373cc15 : SUCCESS in 2h 03m 56s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e8eec050a4ee46fab2de4b4ead2d76b3 : SUCCESS in 1h 37m 22s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e3fef8d8ab7c4da3834f23515564877a : SUCCESS in 1h 29m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/085c1da6ff304feabca7a8ea970a3525 : FAILURE in 45m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fb0ecaf03e1a496b827f325a3253c884 : FAILURE in 44m 33s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e9aadb352bf34ba2b99ed082f905647e : FAILURE in 56m 30s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5f58ab615f864f84b05d44a1bc5392da : FAILURE in 49m 27s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/37dbc81e53cc4ee1a46bdf98ad2768ef : FAILURE in 42m 55s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/f9ea74a669894960bed1a4cd04fe6ea0 : FAILURE in 44m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02580cafba954ea59f51726ad824feb9 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e1e57cb3a39dced53a4622f0688cd232ed13add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 04:41:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"419f72d26b830b1728789e6f61078efdb0deb560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 06:36:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9a567635ad4449eb1f136749d14a52d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f971a998728b439ca6f93fae13a372aa : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ca24e49b14d4221b435c109484de2dd : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3bc1612c94e45d2b560e87135cbc916 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46ece3dfb7eb4691b1dc0553b18be335 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b14e10e0f042f7a88bc46625fa9c84 : SUCCESS in 5m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/350213c43f8e4d67b35415282be56060 : SUCCESS in 7m 49s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/8815bcf444524e59a8ffb96c994ca710 : SUCCESS in 4m 40s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8006965ff22f4e14a8943a988dcb509a : FAILURE in 46m 24s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/716d48cab726477f8593c40feea12cd3 : SUCCESS in 1h 49m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fcb9cb95e1a147d7887d0b3d9d886291 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":1},{"id":"93468dce424de4743d0fadea0e1712e73196275e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-10 07:57:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54b7180b5e6e66acc8e45fbd4db479ed2ea9aa2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 10:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf91b08cac334c168c35ec884e533399 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd87504413d54e9888ced184bad95dd0 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8bb511b0da4557945e871831c664d8 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df27dc00e40c447eb25e4da68d7c493a : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29acdfdc49b4430da2eb39e6c0298740 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f104021c32a448dd9a9107fb1e44d05f : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ec5e3013ed4655bab96db0bbde70c5 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9207c1120a0435cba4c6923ab66ee0b : SUCCESS in 7m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2dc4fd4019d493aad35204060e4ac99 : SUCCESS in 5m 17s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/e40f10700539456180ca480ba75138b5 : SUCCESS in 5m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b6b03f317867495d9dfb3d707d7ca095 : SUCCESS in 2h 20m 34s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/675eb914a3e040b2a18b2d5a23e4180f : SUCCESS in 1h 49m 29s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/47fdb58f20e94ab999b674980714bddb : SUCCESS in 1h 34m 53s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c573419c83eb4544ae3edf486c7a33a8 : FAILURE in 1h 00m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8368a91fc2e4450e9e895fa69f659863 : FAILURE in 43m 00s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7150cd188ca7400bae6d6bfb4964299f : FAILURE in 1h 19m 07s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4bb7db5ba71344c58ae72f623a3205c3 : FAILURE in 57m 32s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/db78480f3e484810bb749631cd3da731 : POST_FAILURE in 50m 19s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/3c197d7611fb4096a66af0875677c1cc : FAILURE in 58m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/361cf828531d475b85d7565167698f7b : SUCCESS in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"164db48382d8bf05cd3130d31325a7878bdeef6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 10:23:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b9f614e4886627ae7ad332f8aca3d22ec410d39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:08:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8d8ddfce6b8440fae109837056dfceb : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5058eed6ddef4b81a8b200e3b71ec01d : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e32d88b3ffb4789913bf7c4ffd55894 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3700a1ab7f0a4b5a91d0c53fc669f568 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb45d96f19b4418a91cc47bb995326c9 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd70ccb50c164ebe9a48885c7077fe33 : SUCCESS in 4m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72e5e9971f164ed1b263db8529d8a8be : SUCCESS in 8m 16s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c6e548d3915f41d9afa799c6aec64991 : SUCCESS in 5m 09s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2449f7e937e94c70b4a2bab77b490838 : SUCCESS in 2h 43m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cc28be99acda4869bb3f016bc7c1a6ee : SUCCESS in 1h 52m 42s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6818e75e7cee4cc5a18689b6c393c72d : SUCCESS in 2h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9ac3a7b99e8411e3611b23c175b13266319f511f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d13dee866e94f541b55bc198ab47803cef9497ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea8ad5810c24a1aa0622a35e8dfd5ed : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bfa510194e3cf30c8c03446d435edd87c1b4d23","revisions":{"6bfa510194e3cf30c8c03446d435edd87c1b4d23":{"kind":"REWORK","_number":1,"created":"2019-10-09 09:21:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/07/687507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/687507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/687507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/687507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/687507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/687507/1"}}},"commit":{"parents":[{"commit":"e45b6a63490649ac0b3fbce27e7c79d547814840","subject":"Use correct Swift credential for instance backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e45b6a63490649ac0b3fbce27e7c79d547814840"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-03 03:48:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-09 09:21:23.000000000","tz":0},"subject":"Remove all the resources when the Nova VM creation failed","message":"Remove all the resources when the Nova VM creation failed\n\nBackport candidate for stable/train\n\nStory: 2006664\nTask:  36926\n\nChange-Id: If0991e0cef40ca78752fcd509d8438e90c9557bc\n(cherry picked from commit 650794eaf9f328bd1b2fc096992e6190a9d79c87)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6bfa510194e3cf30c8c03446d435edd87c1b4d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6bfa510194e3cf30c8c03446d435edd87c1b4d23"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
