)]}'
{"id":"openstack%2Ftrove~688005","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I4d2fe38fac34c75b42234fa7a20aba51e6dc1c4e","project":"openstack/trove","branch":"stable/train","hashtags":[],"change_id":"I4d2fe38fac34c75b42234fa7a20aba51e6dc1c4e","subject":"Support to specify branch when building the image","status":"MERGED","created":"2019-10-10 23:08:27.000000000","updated":"2019-10-11 04:27:48.000000000","submitted":"2019-10-11 04:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688005-1570767999545-8e81ebfc","meta_rev_id":"2c8769f6290ee59ee9e18983b7701d3ee1ebeb1e","_number":688005,"virtual_id_number":688005,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 04:26:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-10-11 01:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-10-11 01:49:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 04:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfc3f8d86648a64059b978eb376de3abcb0dd27e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-10 23:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"202f1a7999c0ccb5bbd0636d8e70003cc6963904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 01:46:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9adcef629294a8da32cf87f6cc3ed63 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e96c7266ebc41afaa5afe80d187e4fd : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474601d1db8d4b97a48b5f759395ad35 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93865725d489471e814b02a7768e4729 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b79cce52c8438ea8fcc6d16e66be68 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c82c92df683418697f32b3e1c965768 : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3ffb322a9648b59bea0dd40cad4325 : SUCCESS in 6m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2bf306acc8e844589117fdb14e419e13 : SUCCESS in 9m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6107ab501994eabb54299b3b395ed31 : SUCCESS in 5m 22s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/634de64244f546c2a66248f8e0102ae6 : SUCCESS in 6m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b60215d6e13a4582b5969126e2bad733 : SUCCESS in 2h 36m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/554b3883244a4d738575118fea323b44 : SUCCESS in 1h 35m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/83bdfecfece14e9f9e210b7b8d9e4075 : SUCCESS in 1h 45m 56s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2b7093f6d45142bdb7dfdd9bad0538be : FAILURE in 41m 44s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/287babb39b6241cdbf3db791bfae6c9d : FAILURE in 44m 44s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7e03b96c7d894a66a3057f69c832694a : FAILURE in 1h 18m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/a7a6184ebc7b42dd854224c9c37c2c04 : FAILURE in 59m 44s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/47b4ed33c9f34e5baddc56ca4a3b5fa1 : FAILURE in 42m 52s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/4026951e22fa44e18b6674879c97ffc8 : FAILURE in 44m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a665e53ce80843c4829f6e30c30c6821 : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54919803fd2f02d92120a6dba612e413ff36df33","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-11 01:49:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1214fb8fd62a131b0c1f99e34187e3902a8d422b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 01:49:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8eedc935c34b06938539cf04536dab7b582bac25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 04:26:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/690007e3ef8148469747db5a4e7bb796 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d427068afa44c478e1fb46e5309c2ad : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4d8e285bba14391a01cd29ed4a36f38 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d2119ba966b4f59bc2f7ffd3f5a095c : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e6a12bc637541be9f4168fd40c93782 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b3225e94724be7b2db3866bfd80f64 : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8929d52a9a5a4d169c59d4a3f12889bc : SUCCESS in 7m 52s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/073bbc7ed1284d7fbc4897fb715355d0 : SUCCESS in 4m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/434c042a8c704965a8147d5cb2072190 : SUCCESS in 2h 31m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c43d316fd5f34853aa150d7bd80f72b3 : SUCCESS in 1h 38m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/42cf3d42fa694d598b514889fbc43d63 : SUCCESS in 1h 44m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a315a455ec9f6ec251cf683b5e8ab21de91cf295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 04:26:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c8769f6290ee59ee9e18983b7701d3ee1ebeb1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 04:27:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f5c9ab015546ef9ae403ff98d17247 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8217b99705e43945aa03ca7b559cc8f1ab00102","revisions":{"f8217b99705e43945aa03ca7b559cc8f1ab00102":{"kind":"REWORK","_number":1,"created":"2019-10-10 23:08:27.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/05/688005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/688005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/688005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/688005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/688005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/688005/1"}}},"commit":{"parents":[{"commit":"e8a6a60f24abc76b5b040fe3dee916865b9623d4","subject":"Fix the stable branch image build job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8a6a60f24abc76b5b040fe3dee916865b9623d4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-05 00:11:20.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-10 23:07:48.000000000","tz":780},"subject":"Support to specify branch when building the image","message":"Support to specify branch when building the image\n\n- Explicity specify branch name for building non-dev image\n- Use branch name for building dev image in devstack\n- Update the doc\n\nChange-Id: I4d2fe38fac34c75b42234fa7a20aba51e6dc1c4e\n(cherry picked from commit d244c9d6661a4595533fcbb972798771791f4ed5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8217b99705e43945aa03ca7b559cc8f1ab00102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8217b99705e43945aa03ca7b559cc8f1ab00102"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
