)]}'
{"id":"openstack%2Ftrove~689118","triplet_id":"openstack%2Ftrove~master~I3401ab18faf4ccdd5006f8258d67626f49aeec5e","project":"openstack/trove","branch":"master","topic":"fix_mariadb","hashtags":[],"change_id":"I3401ab18faf4ccdd5006f8258d67626f49aeec5e","subject":"Fix Mariadb issue without password and innodb plugin","status":"ABANDONED","created":"2019-10-17 10:06:18.000000000","updated":"2019-12-05 09:36:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e25be03092a709b9cf9d7529abff70756b80f2b9","_number":689118,"virtual_id_number":689118,"owner":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-05 09:36:19.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-17 12:59:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 12:59:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 09:36:19.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"67e00076f34f6d781af5fcfd1cd9c18780c817fe","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-10-17 10:06:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5676dffffe2f10c04f202de20fb866676df603ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:59:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df8ca08eb2b4388864f57904e0e3e31 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89a7803446d24cda8f540f8c25a22f7b : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83659b6f2f9f4f6db678a0cf3fb85a38 : SUCCESS in 9m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce82793638804920aadb7e15ba508285 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dde5586acc1e44dfacfd942f3bd226bb : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14ced49a779f415bab8835a02e36bebc : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789ce3bfe3b64bb7b7b52bdced3f2f18 : SUCCESS in 7m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5f0bf32fca24d3da0f41718c8c9ed08 : SUCCESS in 9m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d92ff8a69cd4f498d6520468aed1daf : SUCCESS in 5m 12s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/5a58be8370ad407b8d11cfd93e084e2d : SUCCESS in 5m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/39d31212d38546e0a30967fd36ec158f : SUCCESS in 2h 51m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/00fd87eccdfc448eaa41ab60b21b4eb3 : SUCCESS in 1h 51m 13s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fef0291d92394eb49555abd82c8f26da : SUCCESS in 1h 55m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a0b6ff6127c14993b543092091be7aed : FAILURE in 41m 17s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/873bd1b6fb9b463cac32be7ab335dd30 : FAILURE in 41m 30s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/047f964d45c940b89d999100784657db : FAILURE in 1h 43m 08s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/db9ad96ccee7468c85774ef756bf2ba5 : FAILURE in 1h 22m 44s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/dfcdf5a96b244b738aa8b9bccfb4a775 : FAILURE in 44m 58s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/5bb4e425e35840909b85149dd9a1e78f : FAILURE in 50m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3732b36a079b4515962c24dec5f9980e : SUCCESS in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0a808510f7ec9aab8ace702b740f282e0eb6151","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-04 01:18:46.000000000","message":"Patch Set 1:\n\nHi, I\u0027m working on fixing MariaDB 10.4 CI here https://review.opendev.org/696770, do you mind if I include your change here in that patch?","accounts_in_message":[],"_revision_number":1},{"id":"2f5708315749b1ebe86fc45b64af0724dce2464d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-05 09:36:19.000000000","message":"Patch Set 1:\n\nalready fixed in https://review.opendev.org/#/c/696770/","accounts_in_message":[],"_revision_number":1},{"id":"201205591b757eac6a5536a3683b64954da5a9df","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-05 09:36:23.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/696770/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6323aae61df3e0b1315fb988cce0cb8f8408933","revisions":{"f6323aae61df3e0b1315fb988cce0cb8f8408933":{"kind":"REWORK","_number":1,"created":"2019-10-17 10:06:18.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/18/689118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/689118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/689118/1"}}},"commit":{"parents":[{"commit":"bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48","subject":"Merge \"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-10-17 10:01:53.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-10-17 10:01:53.000000000","tz":420},"subject":"Fix Mariadb issue without password and innodb plugin","message":"Fix Mariadb issue without password and innodb plugin\n\nSince mariadb v10.2 then innoDB is default [1]. By default,\nthe mariadb access without password thus we need to set empty\npassword before create new engine with empty password for mariadb.\nThis is a reason is making error during mariadb mechanism.\n\n[1] https://mariadb.com/kb/en/library/using-innodb-instead-of-xtradb/\n\nChange-Id: I3401ab18faf4ccdd5006f8258d67626f49aeec5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6323aae61df3e0b1315fb988cce0cb8f8408933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6323aae61df3e0b1315fb988cce0cb8f8408933"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
