)]}'
{"id":"openstack%2Ftrove~689418","triplet_id":"openstack%2Ftrove~master~I601aff4dfa820a3dd7c2e0ae6069689e6853d251","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I601aff4dfa820a3dd7c2e0ae6069689e6853d251","subject":"Bump the openstackdocstheme extension to 1.20","status":"ABANDONED","created":"2019-10-18 10:30:37.000000000","updated":"2020-04-24 01:36:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c180731f3e1243dfbc10c947c9b964d789b54a1","_number":689418,"virtual_id_number":689418,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-18 13:01:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"all":[{"value":-1,"date":"2019-10-19 16:48:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 13:01:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-19 16:48:01.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"}],"messages":[{"id":"4cdbf6fa638d8a7443834827e151ebce4672fca2","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-18 10:30:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"975682a12024e3ae4a2fe27eb264306062bc1884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 13:01:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2363cc7ec94d4ffab7ea9a6f40c92497 : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1b17b8a2e354dd39e0f29cec2f0c919 : FAILURE in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1f1534c1e18411283cde1a220f32756 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d75acc4ec5c4759b24c54421f25af80 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad72f5f636c42bf9a5178b79b8afb65 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ecdb00b61044b21b8ee40e0ffd9261d : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb12b8d438b244a0a7bffc7b563e6803 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9752d070be13416ead4c6b956c15d6be : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b82003ef5704405a193051fcdd32715 : FAILURE in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b91500b0a86349ef91422e918f4f638f : SUCCESS in 7m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0e4cbe6679c4d18ba4bfed3c5889091 : SUCCESS in 8m 42s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/fb44705c9098411ea816b8b425f425d3 : SUCCESS in 5m 43s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/184948a4d7cd4497b55b9bfe23a1b273 : SUCCESS in 2h 24m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/eb0283ea2c324e1aa4fb86af279663be : SUCCESS in 1h 44m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7367466198134b4baf8b75abc6b6cf69 : SUCCESS in 1h 40m 55s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/213071b6e71245e38ec01f2b896787f3 : FAILURE in 52m 11s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9a30087dff2a4016af9cb51facec72fc : FAILURE in 42m 25s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2de18a4952134585ae925c36642c4299 : FAILURE in 1h 02m 22s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3fe32af3ba0c4e41abb3478d1e8f1bad : FAILURE in 44m 28s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/b74ae15e565847e6bcfea52e52007aa2 : FAILURE in 48m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/64ff720e475a4171a9e5ddd1f88add64 : FAILURE in 43m 13s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eb89c8edc7942d19d8dea030eee1264 : SUCCESS in 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b513a1c7932c1302859f3a78c4c1bf911062cf58","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-10-19 16:48:01.000000000","message":"Patch Set 1: Code-Review-1\n\nA lot of things you need to add into conf.py in doc/ folder, please following link: https://docs.openstack.org/openstackdocstheme/latest/#using-the-theme.\n\nOtherview, you need to change on lower-constraints.txt to match their requirement.","accounts_in_message":[],"_revision_number":1},{"id":"7c180731f3e1243dfbc10c947c9b964d789b54a1","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-04-24 01:36:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3acaf6d8f02edffa284a7a25e1866e56f4a8962","revisions":{"f3acaf6d8f02edffa284a7a25e1866e56f4a8962":{"kind":"REWORK","_number":1,"created":"2019-10-18 10:30:37.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/18/689418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/689418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/689418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/689418/1"}}},"commit":{"parents":[{"commit":"bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48","subject":"Merge \"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-10-18 10:30:09.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-10-18 10:30:09.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: I601aff4dfa820a3dd7c2e0ae6069689e6853d251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3acaf6d8f02edffa284a7a25e1866e56f4a8962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f3acaf6d8f02edffa284a7a25e1866e56f4a8962"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
