)]}'
{"id":"openstack%2Ftrove~689994","triplet_id":"openstack%2Ftrove~master~I148b50c4fae70cd7155b31cb8e5ebc4f09944d0b","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I148b50c4fae70cd7155b31cb8e5ebc4f09944d0b","subject":"Switch to Ussuri jobs","status":"ABANDONED","created":"2019-10-22 07:34:12.000000000","updated":"2020-05-27 10:56:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a82b15ba65f469d4b38b86c9795c321e921f7c08","_number":689994,"virtual_id_number":689994,"owner":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-03 06:20:07.000000000","_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-22 10:17:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 10:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"361b06a7f416da52072c2e0b9f4f52cf89f37bda","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-22 07:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"662e7f4ed456e55116e6fcd1dd0dc7be62cbd43d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 10:17:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f6462bfaca0454abcdfa82e839eb485 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b38563a44cdc48cf91e38f98b07c77d1 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97ffd4a7e2c43cc98b14455ed27b84e : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3b340d39f104b9fa5db417ab52f18b3 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f397d8256f141b4a17f24029aed98f7 : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2ce1401c058474dacb3a07ca7cec0db : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8671d41007a24afcba008b36cec21abb : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36fc2c460a55432e9bb0a576954311fa : SUCCESS in 9m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cf7c549ef9646868e7e6049dfbd9d22 : SUCCESS in 5m 16s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/7cc4b836172846ffa3e54c2c83921722 : SUCCESS in 6m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fbb1abad9c4240a089363f0be86948e5 : SUCCESS in 2h 34m 36s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2a3a1a2e48384df88aca0ae82ac6dc52 : FAILURE in 2h 19m 11s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/8171f2d71afc4edabb019fc26c936469 : SUCCESS in 1h 54m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/366e9f78f5de4028a47a3bab9bce6b86 : FAILURE in 53m 06s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7e83ed7394044b25a904ac1ce7901201 : FAILURE in 47m 51s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/67229d36d53743b3b80d023edb7b0783 : FAILURE in 59m 39s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ac725de005de4c338c12f0e8e6173d09 : FAILURE in 1h 00m 34s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/45902a0267f948aea466313bbd459da7 : FAILURE in 44m 16s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/9af95aee00834e698c260c5caef0b69c : FAILURE in 45m 12s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee6641dda63f477fb7e1af11f798d483 : SUCCESS in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e5fd3dad14b68b748c1a28d215d28ebf8e3a97d","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-12-03 06:20:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4dc07490fb4e675750b1469f6e3f5c74e96ffea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 08:11:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a7dab2690614e149c23c0cf0b4a4ee9 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d9ee18dfe4e454f921729abb382e251 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab28e82405c437382381599b4444899 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b2157d7d7b845399359ff10a883cd0c : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d776e4d6d641f9b55ff1ffe66b9803 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c39c91b15944bafb4dd3d1bd77ae41b : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6abe5263202451bb082507b1b84c327 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/340d476803864606ad69cc6aae83f51d : SUCCESS in 9m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5db3c7d8833147bbbe91bfe2bfae8824 : SUCCESS in 7m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/51e54819362f4420a1360a874e073162 : SUCCESS in 8m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/624f406de16849eaba7561a688cd8914 : FAILURE in 1h 17m 44s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2ec2fbe827a44ad28d73b612f617798d : FAILURE in 1h 40m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/30341fd52b364d95a6800be19efb7f56 : FAILURE in 1h 06m 12s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/bb10f0fd61d64f7c913b550448269d69 : FAILURE in 53m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e3b61ba8ee5146489658cad2dbddf51e : FAILURE in 52m 56s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/38f8695075f24a539f9c1706d2001d95 : FAILURE in 53m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d2dff16535744b31bfd5edf9b27224d7 : FAILURE in 51m 59s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/10d117831a8045b2ac14a81f843bfecd : FAILURE in 50m 43s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/3a13b2fbb3464165bdf05fc33d4412e0 : FAILURE in 45m 35s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82080a1b246b414d8eae661633610aa0 : SUCCESS in 57m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a82b15ba65f469d4b38b86c9795c321e921f7c08","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 10:56:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f55772ec2721accb201345bb9be9049c3e9bd85","revisions":{"2f55772ec2721accb201345bb9be9049c3e9bd85":{"kind":"REWORK","_number":1,"created":"2019-10-22 07:34:12.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/94/689994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/94/689994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/689994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/689994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/689994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/94/689994/1"}}},"commit":{"parents":[{"commit":"bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48","subject":"Merge \"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-22 07:33:04.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-22 07:34:04.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I148b50c4fae70cd7155b31cb8e5ebc4f09944d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f55772ec2721accb201345bb9be9049c3e9bd85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f55772ec2721accb201345bb9be9049c3e9bd85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
