)]}'
{"id":"openstack%2Ftrove~690012","triplet_id":"openstack%2Ftrove~stable%2Fqueens~I76df1464d7bd7af76afa619766c1721afc1d169e","project":"openstack/trove","branch":"stable/queens","topic":"fix-trove-ci","hashtags":[],"change_id":"I76df1464d7bd7af76afa619766c1721afc1d169e","subject":"Add local bindep.txt","status":"MERGED","created":"2019-10-22 08:53:21.000000000","updated":"2019-10-26 10:31:16.000000000","submitted":"2019-10-26 10:29:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690012-1572085785576-cc9bba01","meta_rev_id":"52c31b44470077d75a5f6bab59c5fd5eb1d5b7ef","_number":690012,"virtual_id_number":690012,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-22 09:42:41.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 10:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-10-26 08:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-10-26 08:30:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-26 08:30:58.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-10-26 10:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9510e94343715b66db054031fa3e4b97038f74da","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 08:53:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c428878f296a90b73a96ef8ff2064c3552fe19b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 09:39:41.000000000","message":"Topic set to fix-trove-ci","accounts_in_message":[],"_revision_number":1},{"id":"bbc011473b1ed8c6561b47bb4c96ac47a938c70b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 09:42:41.000000000","message":"Patch Set 1:\n\nStable periodic jobs fail [1] (and also check and gate jobs), this patch fixes those errors.\n\n[1] http://lists.openstack.org/pipermail/openstack-stable-maint/2019-October/thread.html#15584","accounts_in_message":[],"_revision_number":1},{"id":"114ce04c7b6584998d245de6d227e9427b7cc1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 11:07:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a517fbfca3a46cd81218656389ce525 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/910e927c7356483a8c54c9fa9da0ab98 : SUCCESS in 13m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6134d5f648ca4bc4bc7992090c42b858 : SUCCESS in 15m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/63f88349355643a59bced343037a9f06 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e64b717557dc417199909cb25b6bd58c : SUCCESS in 7m 22s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e163a7a17f8e4ecfa70ec971c1982ab0 : SUCCESS in 1h 53m 53s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/035be46d032043ee83bb8f4961aafad0 : SUCCESS in 1h 35m 39s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/79259fe3ca7c4c8bac1bf4b475b232f0 : SUCCESS in 1h 23m 43s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b1e861cdb7354f5bafefdf868516541b : SUCCESS in 1h 59m 34s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7a82e541804a466ca32a5bd639062add : SUCCESS in 1h 33m 40s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/15db56340a4544e08774ba39413c2122 : FAILURE in 1h 41m 33s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/41b249b2181c4656bfe2c20e347f7cf4 : FAILURE in 1h 08m 40s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/9529e500b8fd42a1bcdb104924d62fc5 : SUCCESS in 2h 08m 03s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/554a67055f2a407e92ebb27b6c0cece4 : SUCCESS in 41m 56s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9f24659cf5064a719597195098ba724d : FAILURE in 1h 28m 12s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/1987fb68eb5d478b9d1aadae49320768 : SUCCESS in 1h 16m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/e68083f399984935a03dc5f15a75f71f : SUCCESS in 4m 40s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/a3916fb954dd486e9c491ad964f6fff2 : SUCCESS in 4m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/645a267c3b204445bfe3bdcd4c106303 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b17872e5c640afe7059d3fd5223b67557fd74738","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-26 08:30:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b62e81979bc6f38df0b608d24301462dcc2f303a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 08:31:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3e033df1f0fc469b594e8be17bc5a73ba5efa59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:29:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b8d8703cd048838f50b335b2536c96 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78a059564a844e63b2cf8e0ed8196acd : SUCCESS in 10m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d2887bff2f51405aa8fa6f55aff0aba2 : SUCCESS in 12m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d36533f6291144529f4690084df0eb15 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86661cfcf27047afb31318e5486ac3d6 : SUCCESS in 6m 20s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/da33969446e24930b41e78eb149016e8 : SUCCESS in 1h 31m 52s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/338a6b98d32840589ac6783b0a09ab43 : SUCCESS in 1h 52m 55s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f2a9e5c6b02f4818bcd53c7eff443c21 : SUCCESS in 1h 35m 24s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/1bb1864ce2fb48919d581abb1bfe0964 : SUCCESS in 5m 05s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/251082868472418aa6d890573f307562 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3867603f6a7a4838ac065338b4e3dd7e : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1f051786442440a37e6e79279877130029fc35cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:29:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52c31b44470077d75a5f6bab59c5fd5eb1d5b7ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:31:16.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a1a90875544ebc818b0a2038793bc3 : FAILURE in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bac67e878479c8edb34a952c850eb8b78d0f8e0c","revisions":{"bac67e878479c8edb34a952c850eb8b78d0f8e0c":{"kind":"REWORK","_number":1,"created":"2019-10-22 08:53:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/12/690012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/12/690012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/690012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/690012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/12/690012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/12/690012/1"}}},"commit":{"parents":[{"commit":"43093806314d234547e0e50fc3775e6b4e54acb5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43093806314d234547e0e50fc3775e6b4e54acb5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-21 14:53:51.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-22 08:49:03.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I76df1464d7bd7af76afa619766c1721afc1d169e\n(cherry picked from commit 239f9c6e4fa7d7bb9bab396f79336725a149126c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bac67e878479c8edb34a952c850eb8b78d0f8e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bac67e878479c8edb34a952c850eb8b78d0f8e0c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
