)]}'
{"id":"openstack%2Ftrove~690031","triplet_id":"openstack%2Ftrove~stable%2Focata~I76df1464d7bd7af76afa619766c1721afc1d169e","project":"openstack/trove","branch":"stable/ocata","topic":"fix-trove-ci","hashtags":[],"change_id":"I76df1464d7bd7af76afa619766c1721afc1d169e","subject":"Add local bindep.txt","status":"MERGED","created":"2019-10-22 10:02:52.000000000","updated":"2019-10-26 10:31:11.000000000","submitted":"2019-10-26 10:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690031-1572085787303-df6c426c","meta_rev_id":"ddd2423956deeee865b2ea387a038872625a10f1","_number":690031,"virtual_id_number":690031,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 10:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-26 08:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-26 08:31:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-26 08:31:30.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-10-26 10:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51c374ce71673228edbca91035f69c4a1bbe2afd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 10:02:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36c33f588f728e7a21df26dd93a5020939f42870","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 10:03:46.000000000","message":"Topic set to fix-trove-ci","accounts_in_message":[],"_revision_number":1},{"id":"6025c1147b81ee47ca3e53c3ee7450a589627b46","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 10:04:39.000000000","message":"Patch Set 1:\n\nStable periodic jobs fail [1] (and also check and gate jobs), this patch fixes those errors.\n\n[1] http://lists.openstack.org/pipermail/openstack-stable-maint/2019-October/thread.html#15582","accounts_in_message":[],"_revision_number":1},{"id":"cb3ad4980d76f71cb469412c940820b88d439b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 10:28:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-install-guide https://zuul.opendev.org/t/openstack/build/09cccc402f4f489a92d5f4e3202763dd : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dfa12350264f42a2b2807bfd33b1f2 : FAILURE in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89e3f6623f8d4564b3207dbe4d37d157 : SUCCESS in 10m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9026b7b37bed4f7fb24748022da441c3 : SUCCESS in 10m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a025beb79cd64629879b08067a3c0256 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fa1d338bd5c4e48aede6ae4c4be52a5 : SUCCESS in 6m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13199c10f4b8426c9b78938593b49a13 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f1f3c50627bd2c02f86beb84a9d5c8831391adcc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-22 13:04:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b94292e54b98be72cc8b5556c4bf3595b5805131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 13:26:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-install-guide https://zuul.opendev.org/t/openstack/build/0d2012fef7f644508bf436a4ea5802ca : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7ffd8c59bd4490b2f56320fe8e1049 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e4f7efc851845b08fe740bf626d16aa : SUCCESS in 8m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/01fc18ba4c2548248678accedfc5f00d : SUCCESS in 11m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d6880f6e92ec4039b5eef395c019d3bb : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/051074d70c0f41fd8733006bd5e6f1f5 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a58c41a94ce44f8cb5d2ea5899637f3f : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3f882137abddc37b4d6b513fd2cfebc368aafdda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-26 08:31:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"39329007059a3f8498a89be74969d8ef203fe7d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 08:31:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"baabed28a0b6bd63e7655474c0d9dc09f2caa757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f0cd69df046b6b22a07655a99b0a2fbb918e93da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:29:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-install-guide https://zuul.opendev.org/t/openstack/build/85ad342b3bbf4a149cbac95aaead3e42 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1715a32cf555409a8d4b4cb187687a67 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ab62f111c3b43458bf3b42d58e5987b : SUCCESS in 9m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4207d35ae7e4c0faeb57050fb36b603 : SUCCESS in 10m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7ae3698dd0f24521947869116e620250 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbbef073317f460ca1f00c283e927f41 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a3b62caa1bd4f3eadc8b9c6a8e4a9bf : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ddd2423956deeee865b2ea387a038872625a10f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 10:31:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7923078b92444193ace9ac5379659495 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12ba44ad3654a51c41771cfd7c3491539e0fcf31","revisions":{"a4f441f2540602f394838563bb38be185a07dd08":{"kind":"REWORK","_number":1,"created":"2019-10-22 10:02:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/31/690031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/690031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/690031/1"}}},"commit":{"parents":[{"commit":"99224990d7f2c50b2700ac8849104099800f9c33","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99224990d7f2c50b2700ac8849104099800f9c33"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-21 14:53:51.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-22 10:02:27.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I76df1464d7bd7af76afa619766c1721afc1d169e\n(cherry picked from commit 239f9c6e4fa7d7bb9bab396f79336725a149126c)\n(cherry picked from commit bac67e878479c8edb34a952c850eb8b78d0f8e0c)\n(cherry picked from commit 908cdb0eae95501963e0da62c7576b3a83c91916)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4f441f2540602f394838563bb38be185a07dd08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4f441f2540602f394838563bb38be185a07dd08"}]},"branch":"refs/heads/stable/ocata"},"12ba44ad3654a51c41771cfd7c3491539e0fcf31":{"kind":"REWORK","_number":2,"created":"2019-10-22 13:04:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/31/690031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/690031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/690031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/690031/2"}}},"commit":{"parents":[{"commit":"99224990d7f2c50b2700ac8849104099800f9c33","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99224990d7f2c50b2700ac8849104099800f9c33"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-21 14:53:51.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-10-22 13:02:45.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I76df1464d7bd7af76afa619766c1721afc1d169e\n(cherry picked from commit 239f9c6e4fa7d7bb9bab396f79336725a149126c)\n(cherry picked from commit bac67e878479c8edb34a952c850eb8b78d0f8e0c)\n(cherry picked from commit 08027d0ff997c02cb8ee4cd2ececd43b2712857e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
