)]}'
{"id":"openstack%2Ftrove~695116","triplet_id":"openstack%2Ftrove~master~I6a06926f833d11691e0995f5baf373a25327564e","project":"openstack/trove","branch":"master","topic":"fix-devstack-image-build-postgres","hashtags":[],"change_id":"I6a06926f833d11691e0995f5baf373a25327564e","subject":"Fix devstack guest image build for Postgresql.","status":"ABANDONED","created":"2019-11-20 00:01:15.000000000","updated":"2026-09-17 17:24:19.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6579bab4d0dbf2d5f7d6c9c3e983aa8dc4cec5e","_number":695116,"virtual_id_number":695116,"owner":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-20 03:05:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"all":[{"value":-1,"date":"2019-11-29 17:16:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 03:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-29 17:16:09.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"}],"messages":[{"id":"5e93179633417debb88b1a38937a7fab4538b461","author":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"date":"2019-11-20 00:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3345e4ecc018d7bb2949b66cb94275d3d7cf9ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 03:05:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c870952a9d4881ac0273256679c474 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acda2f97c34c47f8bbaca58299d80b60 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e600d21336846c08f504e1b6de1e822 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c910e22effb4e6c8a1b5e8da0bfca53 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8ad122bb3e9475a8431d467bfc8364d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16c6815bc6f345eb830b80b55657bf10 : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4998375e9324260b4aa932da4e16ed3 : SUCCESS in 7m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2bd1d8401a084c85ac0339886beb1767 : SUCCESS in 8m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/846da5835d87473abf3e298e5c42707e : SUCCESS in 4m 48s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/a4a5830cf98146388f1d04f561b4f9ad : SUCCESS in 6m 22s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6dc0ef60d14945c39f74232d6a88b8f4 : SUCCESS in 2h 28m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b89d4292b5ea49e3a9404a236cef6d7c : TIMED_OUT in 3h 03m 21s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/74893a2a59e440b8aa01a2accd6f5eab : SUCCESS in 1h 46m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/46db32aebe354e32a62fa22592e303f2 : FAILURE in 1h 54m 43s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/22cbc7dd5a084222af75f1850e9fa1ae : FAILURE in 1h 26m 08s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/82c1d00d116f4f02bca311d36fc34aca : FAILURE in 42m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/bf9e3156205841a3a566866d26e9ff0b : FAILURE in 40m 25s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/1fe7cfd6a7a6470db7fe2202eed921c7 : FAILURE in 38m 01s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/519a3c6cc1c24ed887996b63c551b0ec : FAILURE in 44m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01c88cf142164c2d9f0e5dccc2cf46a4 : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be09266f92d12a4966c7f10e1e6c7221116407ed","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-11-29 17:16:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThank Mark, I have some following comments. You also can create a new element to permit postgresql which will work with guest-agent.service in another commit.","accounts_in_message":[],"_revision_number":1},{"id":"d6579bab4d0dbf2d5f7d6c9c3e983aa8dc4cec5e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:24:19.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3db61f20f081d79425484004b97eb48d2b5964c","revisions":{"d3db61f20f081d79425484004b97eb48d2b5964c":{"kind":"REWORK","_number":1,"created":"2019-11-20 00:01:15.000000000","uploader":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"ref":"refs/changes/16/695116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/695116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/695116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/695116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/695116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/695116/1"}}},"commit":{"parents":[{"commit":"bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48","subject":"Merge \"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdd44790fb10b20a7ef4ceed161bb2ca0e4f5c48"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2019-11-20 00:01:15.000000000","tz":780},"committer":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2019-11-20 00:01:15.000000000","tz":780},"subject":"Fix devstack guest image build for Postgresql.","message":"Fix devstack guest image build for Postgresql.\n\nUse DIB_RELEASE not RELEASE for repo based operations on the guest image.\n\nChange-Id: I6a06926f833d11691e0995f5baf373a25327564e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d3db61f20f081d79425484004b97eb48d2b5964c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d3db61f20f081d79425484004b97eb48d2b5964c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
