)]}'
{"id":"openstack%2Ftrove~697870","triplet_id":"openstack%2Ftrove~master~I6e10744f86f19f3c29babc07fda77936c95022ff","project":"openstack/trove","branch":"master","topic":"devstack","hashtags":[],"change_id":"I6e10744f86f19f3c29babc07fda77936c95022ff","subject":"Move the iptable change from trovestack to devstack","status":"MERGED","created":"2019-12-08 05:51:09.000000000","updated":"2019-12-08 13:32:26.000000000","submitted":"2019-12-08 13:31:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697870-1575811884693-73dee3f6","meta_rev_id":"e5e2c9495d12c52910fb5417aab9458f57defe4c","_number":697870,"virtual_id_number":697870,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-08 13:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-08 10:02:07.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-08 10:02:07.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-08 13:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"061195aed5f051d123bb958cc98d9b156682ac95","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-08 05:51:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33d4df4c2c4b7c1e20f54af531f003f8cf5b0d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 07:23:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f98b20afb2141ceb60aca22e3f509ad : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b95536308b3f487a8a8ed8fe51a9bd68 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be9cf84b8554c7abe38736ea0d1824b : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7540b9c422948f8ae81f48ab779dd10 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5a091af711149af8b72e562e97686cd : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601201e85cb74e548b7df43f261ab701 : SUCCESS in 5m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d0ed6a7af334b7c9892a9c9626b8ca5 : SUCCESS in 7m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/622efacd1bfc4508a7ebe83e00075f64 : SUCCESS in 5m 19s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/67a5566e53d645b892ea43121c694305 : SUCCESS in 6m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/42b3e66391c34cc680528d118415632a : FAILURE in 1h 02m 43s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/aa646052446d4af3b51d49e648a87603 : FAILURE in 1h 03m 20s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2dd0899292b04158a753226628bb42b5 : FAILURE in 1h 28m 52s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d5d77cb21feb44d2911bdc4f710c5dd3 : FAILURE in 1h 02m 51s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/de2973ebfa0b492b8e96e254bf3d1a9f : FAILURE in 40m 42s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/2422f06d5479462ca5e81ac421c89aa6 : FAILURE in 39m 33s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4e3fb35b529b452d869a26f7492012cb : FAILURE in 1h 28m 10s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f4dec712cdac426abc846da5cc1127be : FAILURE in 1h 00m 24s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/7cc47977afdf44a39ddd9b29cd6e5793 : FAILURE in 1h 01m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3abd9df0c3e4daf91fd00afcf586050 : SUCCESS in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d97211048d93e6021e8757c1dc0bdd61b896f672","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-08 07:47:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"90f7a6fc410bdbb7fbdfc34b87ad5afab7b4454a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 10:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22feef6d006d42ebbd15897b8d1ae455 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8ebf3456b4429cb7717ab95b953d04 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf440e8077e45c3b51b70e683407b60 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9c4279fa674577ab90de73c725da54 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd4d839fd2bf45b286e3dd66415d656b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520c91bd94794f95bb91e271d7b5b224 : SUCCESS in 5m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ddee9af368d0402d9d8fe62531c38c8d : SUCCESS in 7m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72f0a83fdbbd484a875dac0437a5aacb : SUCCESS in 5m 21s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/3d906fc09fdc4de387b81a5182661b60 : SUCCESS in 4m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6197a2b915984397bab770f02c7f1c56 : SUCCESS in 2h 09m 06s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/805012968c9041d0a681f4217db06ac7 : SUCCESS in 1h 34m 14s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7998ca9d1bcf413792649052b71519f7 : SUCCESS in 1h 37m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e0b69cbf10fa49539d5107090b4ff6f8 : SUCCESS in 1h 30m 07s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/36a2b21feb3b403e8cb2991acca6b2d5 : FAILURE in 39m 01s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/b42d996d85874b4e8438b2bd233433fe : FAILURE in 42m 16s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/13035823c98140cc9e817044c8b35bd0 : FAILURE in 59m 51s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/246cb2e3c5e74e429c7ec10a2c704480 : FAILURE in 58m 28s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8222d8511b5f4a428b9fe322e484edfb : FAILURE in 58m 55s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8834398d1ace4a648e27a0719ae2cf11 : SUCCESS in 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"765f8ae79c341b34a130d2f21871ca795859b1f6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-08 10:02:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32a05bac470f615ae88b6ac5ae6d0d28f77ae660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 10:02:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a496dbc6bdaa225ca1a8709b53d2940bb8f02a82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 13:31:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b48c3719397248429389ca3377996a70 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518fe2aa94b244ae86b620a9bebe141b : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d706391bcdc74d0b9727a7b5526a86d9 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a611bad0f824c7bac60d899c14c9845 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a80b7e86644beebeea2b3d7d0c95f6 : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0117d9b5928484695f257eec2e42ba6 : SUCCESS in 8m 14s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/86c38045654447d4853c33a208932035 : SUCCESS in 4m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0a4c05473d214b2e9b0a818c3e93db6a : SUCCESS in 2h 50m 11s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e1025c3c751d45c99cc30d3631191a5d : SUCCESS in 1h 39m 21s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ac96d48fd2de48da8cce609811d3aec9 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a69c1136ee1e84e2b81b1776fbb9d80ac5431429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 13:31:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"677e7323e4c34fc91eecd805c49a824e0fa8fcf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 13:32:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7cb72bb595469d94d349695fe638ef : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c551947c9983109d494de92275ca4be2d11510d","revisions":{"4c551947c9983109d494de92275ca4be2d11510d":{"kind":"REWORK","_number":1,"created":"2019-12-08 05:51:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/70/697870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/697870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/697870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/697870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/697870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/697870/1"}}},"commit":{"parents":[{"commit":"e226ba68abfe42a2baae6017c2ae14d38dbd974f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e226ba68abfe42a2baae6017c2ae14d38dbd974f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-08 05:50:42.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-08 05:50:42.000000000","tz":780},"subject":"Move the iptable change from trovestack to devstack","message":"Move the iptable change from trovestack to devstack\n\nChange-Id: I6e10744f86f19f3c29babc07fda77936c95022ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c551947c9983109d494de92275ca4be2d11510d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c551947c9983109d494de92275ca4be2d11510d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
