)]}'
{"id":"openstack%2Ftrove~697954","triplet_id":"openstack%2Ftrove~master~I87f3eba550326abcef1985a001a358125b6820a8","project":"openstack/trove","branch":"master","topic":"trove-tempest-ci-job","hashtags":[],"change_id":"I87f3eba550326abcef1985a001a358125b6820a8","subject":"Add trove-tempest CI job","status":"MERGED","created":"2019-12-09 07:59:09.000000000","updated":"2019-12-09 15:19:10.000000000","submitted":"2019-12-09 15:15:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697954-1575904508596-612455a9","meta_rev_id":"68afb493c7220d07a995587e629b4f303c6aebf0","_number":697954,"virtual_id_number":697954,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 15:15:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-09 12:00:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-09 12:00:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 15:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"459122e431308d1b6e078f6bfb5fbcd3b82a6927","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-09 07:59:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33360382bc3ac5f2e65793dc945835a18bb7b3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 09:42:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0a7aaf330e74b43b90ad87adf51f039 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec7c4275bb24212aa0a11a1aa730143 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac2b93d498f4b6fbe41815125f765b5 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b61b51221db4d74ab920fe4d29ab105 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b03bcd84b19e4febb595b5cbb588ce11 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba2d0d02c054cf9a850da40b6a50c3d : SUCCESS in 5m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db8b8f7f62b44ac697e59ab1f6cdcaf0 : SUCCESS in 8m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8aa37be45aa440d90fd4c0f09be6c7f : SUCCESS in 5m 32s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/8c11dd13bec749a7988682c9a8d4ed86 : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f65c35f8563246b3bbc0abf82e6f56c1 : SUCCESS in 49m 36s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4661f0a10311457692b3a3b940978b9b : FAILURE in 1h 36m 54s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/204fe23718ee43c296e7bee46fad415b : SUCCESS in 1h 32m 04s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/10b684f4ee574bb499b99a6b2304467d : SUCCESS in 1h 37m 25s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/8c84f6e1092e4fadaa7a453217122db2 : SUCCESS in 1h 25m 59s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/01513fcba8e0416ca16bc9554ef2a680 : FAILURE in 36m 58s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5489d55e8a4847e485e5ede9dcd67aae : FAILURE in 41m 42s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7bb3651d59c14e03afc3c6dfadf2fd6a : FAILURE in 1h 00m 59s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/a15af0b5cb8440899a8069951bddc849 : FAILURE in 1h 03m 11s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/ef39c9d09e5e47afa7663b611b7d0f5c : FAILURE in 58m 13s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d839990a03ba4e5b823d50ac4ef8e549 : FAILURE in 1h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03e727bb73795f1fb4e4d5b3eba2d2386dc5de7c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-09 10:07:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9b7984c91c1b35c567e2887dd5cf442ac14e30f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-09 12:00:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be19250f1f4c02559243ca27c2b4159b7d9fc270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:45:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18cab23db8d2489d817d38e61eb1bf36 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a9e5b620479450e84a0e26f48756e1d : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35ecb96d621f43c4a5d5e601e327e996 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf67d22e5f814bd18ed433175bf8675a : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4927e879468841e190ceb627bb8f77d0 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08fd7c6e7934a39b1e25c82ce0321c1 : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/56e45a857b3a4e9aa521ab39b31e4193 : SUCCESS in 8m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c26fb15490a45d98531bbf65222998f : SUCCESS in 6m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/8a44c78c0c59472e8bf5d1711dbcd395 : SUCCESS in 5m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8c7757eb6d6346a78b7e72b504aa9a00 : SUCCESS in 53m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/de339463f38045548a1a5f829befe0a7 : SUCCESS in 2h 31m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6409c82e855a4e4780b981d338241d50 : SUCCESS in 1h 36m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e0484159607f4aac9c2e276b0d4aac7c : SUCCESS in 1h 51m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fda3618ffe0c4ac5bc23cd30daec02d5 : SUCCESS in 1h 39m 04s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/06c1a408795e40df8529b83792d72ea3 : FAILURE in 42m 01s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/dc6f0844fae24650b2093d2aa5c63944 : FAILURE in 41m 48s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e60d5b31dbd44dc28c8fc60269fdaab7 : FAILURE in 1h 06m 09s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/2394680249034de3b240911a9eaa49f7 : FAILURE in 1h 01m 23s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6cdddcb733a1495a8600029980e6630f : FAILURE in 1h 01m 12s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/325eab918bfd402ca91420eb3b762c8d : SUCCESS in 57m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a57df1e5843d807218db6ea436ed5089b7ee3403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"103a339a76bbcdc913ac98dbe28d01a67886d597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:15:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d690fec1255c49b59a43b2f3347283aa : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d27fa1a20849eeb36ad872709ee270 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21db1ce4fd98430291f071eebdbc780d : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6a433ac92f845d9b778d0d55106221e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973bb91a82994b2e952c204ac9ae492d : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9de8268eaeac4c35a3ecb201b8c70931 : SUCCESS in 8m 06s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d44d0ed877b44d9ab51e200cc08a3d59 : SUCCESS in 5m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/280f4275e0e04034a49b36dafb04b3cc : SUCCESS in 2h 26m 28s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a8fd65fb714f496fb522594927452fb0 : SUCCESS in 1h 37m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/bb405c5ad54348d8b367bc31cfd442a7 : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c321f93c1e48a63091494a26b289a1b32ed17034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:15:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b20f1c6b057482d6ac6114fe8cdd7af8e32b8726","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:19:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0fbdc5c07844a7aa51793b51d2c049 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a149be4eb4d307c31958fd310d49f86b6c860d4","revisions":{"5a149be4eb4d307c31958fd310d49f86b6c860d4":{"kind":"REWORK","_number":1,"created":"2019-12-09 07:59:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/697954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/697954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/697954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/697954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/697954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/697954/1"}}},"commit":{"parents":[{"commit":"b2a31c5ac5b2abd902a075fa20db07962511faee","subject":"Merge \"Support incremental backup for MariaDB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b2a31c5ac5b2abd902a075fa20db07962511faee"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-09 01:45:08.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-09 06:38:11.000000000","tz":780},"subject":"Add trove-tempest CI job","message":"Add trove-tempest CI job\n\nChange-Id: I87f3eba550326abcef1985a001a358125b6820a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a149be4eb4d307c31958fd310d49f86b6c860d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a149be4eb4d307c31958fd310d49f86b6c860d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
