)]}'
{"id":"openstack%2Ftrove~698218","triplet_id":"openstack%2Ftrove~master~I2108f2c6bf067ce228829de9c4e7e47005acf358","project":"openstack/trove","branch":"master","topic":"api-doc-quotas","hashtags":[],"change_id":"I2108f2c6bf067ce228829de9c4e7e47005acf358","subject":"Add quotas resource operation in API doc","status":"MERGED","created":"2019-12-10 11:07:39.000000000","updated":"2019-12-12 05:53:44.000000000","submitted":"2019-12-12 05:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698218-1576129942723-a31ee413","meta_rev_id":"0eae6075adddd6482bee2a63fd93b85baa93e935","_number":698218,"virtual_id_number":698218,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 05:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-11 09:31:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-11 09:31:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 05:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d51255bbcc5a884d6482ad04959df46c631849ed","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-10 11:07:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"894ff0ac56cafcf1a73c937b126ab1326309c169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:50:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4ce7b9c83c7340e8ad8eb304d1636a29 : FAILURE in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3218e64421c47aea0acc0826c1f5c9b : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b9a43fd40ff493da375b966b1ad7ba1 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2807d37e2e4fde96a8831b7cdb9e71 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/614a5a0b015b496b863e056716d91297 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41906e5e39934d478d2b7f5f792567a5 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75618e2c6770428ca605113800954286 : SUCCESS in 5m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7725eb0d97c473bb111e015289667e4 : SUCCESS in 8m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/82d47a8370454b2e90a13aa2ee11e0a6 : SUCCESS in 7m 07s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2333f0eec12e4f74b1b10b08fbfaa5b4 : SUCCESS in 5m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/80fefc6d95dd4abba6de7e1fd98c316e : SUCCESS in 58m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c95ee64b7c294f5e9fad02541490eb09 : FAILURE in 2h 36m 25s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0c28452d75fd46d09418923f22aac04b : FAILURE in 1h 18m 04s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d94d8ba464c2470fb106efddef5988cf : SUCCESS in 1h 41m 44s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/cad51819e8c24769ba6ed068bae475a4 : SUCCESS in 1h 42m 53s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ee52578e50e0432aaa1939e9b08fb953 : FAILURE in 45m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ac0538366e934da383305a0f83f4887a : FAILURE in 41m 52s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0ee0328647e449418cdee10ab78bec3a : FAILURE in 1h 02m 45s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f8147db1986342009acf272873fb68ad : FAILURE in 1h 09m 56s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6d36bf5cd4334a0b87899cf3eb1c5f61 : FAILURE in 1h 03m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91debc0b204447c99f32e49e1f55e9e6 : SUCCESS in 52m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41cbf7edad7f14f51805062adcdcfd4be866152b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-10 20:07:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21f5ed2f149e215a74d6a2f18aaee9aaec4d1750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 21:56:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/99c095dfdf964322ad3ff36802655618 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79da0f59a5604dbb8d23dac75ad7509c : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d545e4bc3b784ecc8d8b4918178c6122 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b131cc895142a6ac42abd89a250444 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7f19d9e3a742339ebd3e4933216e00 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cd0e62f30fd40528acc57ee97c9ad13 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09db61dd94024f40b7d5162c9f932148 : SUCCESS in 4m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f54626b1b954646b287b711e69ac722 : SUCCESS in 9m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37bb630189c04165a35ecaf53bfdf247 : SUCCESS in 5m 40s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6c54cf73bbf34d18a17bb05bf57dfff6 : SUCCESS in 7m 49s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e591804819a644b78151a64f7ec07706 : SUCCESS in 1h 08m 59s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/eb1f4ad0d95e4219aad0eeb9cf01619c : FAILURE in 1h 08m 06s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/abfb0b043bbe4f52ac8071eb6204a946 : FAILURE in 1h 17m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/aabcceb72de64249ab154df888c780f1 : SUCCESS in 1h 36m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/42d23d6d893f42edac2b72810f659114 : FAILURE in 1h 26m 46s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ea5897b093784ca7bc0d1ed15474ac01 : FAILURE in 40m 03s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4ca84c106c7b43a4a1a6a610c6db31e8 : FAILURE in 40m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ab6caa8a86094d898818e73de57860ec : FAILURE in 1h 05m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/64f0f6650ee241d7932fc500900c1172 : FAILURE in 57m 12s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/949415c99741418ebd5d7b85d58bf406 : FAILURE in 1h 13m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c74239e485b41c1868ff3a6f4a7a439 : SUCCESS in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f3afb15cac909eb58efdeb8218ce1df25103584","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-11 01:55:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"897dd00aa61905a87669715469ea6cadf7b23dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 04:11:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d540782875d4cc4a5ce2d9022b5e002 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34bddffdebc34eae93676d5c1bd24848 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63b523ce7f04eaa847557ac87cd4eb9 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cac94e0fe914a4381ccb75ff904b35e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/122d6afd41744a7093ca99478b318f5c : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acfa3937ab4f436b92b1be6eaa62d486 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aeb6e1a268547359a07472844a4a113 : SUCCESS in 4m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecaf0977b95b4254a96aee08ddad2227 : SUCCESS in 7m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce7ff4df20934f1895495976b2150194 : SUCCESS in 4m 55s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/cf764ba760f04a6c86dbc0fc051b8cdb : SUCCESS in 5m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0036268af7824377bc096e489c2d52a3 : SUCCESS in 53m 59s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/616be64a138448f48378d5569cb08a4e : SUCCESS in 2h 14m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6f0b1364764f414eaa310b370fdfd829 : SUCCESS in 1h 42m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/24524c4769c8472ba6cc47358ab651ad : SUCCESS in 1h 25m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c15b0d53e65c43e3950932a050001d4e : SUCCESS in 2h 00m 34s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/52764aca4b734e79afa98c2c0de00d52 : FAILURE in 36m 46s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e9b282c55bb541b68ac3cd604ec25d4b : FAILURE in 39m 09s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ae74ab83a5bd44f69555bf54f0dbe96e : FAILURE in 1h 09m 09s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/f24731c4720546a4a246708906bca6ca : FAILURE in 59m 33s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/9707f2f0985d49eaa926a9470808963a : FAILURE in 57m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/878a3ad66b3b45d086672380560e1818 : SUCCESS in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44de3a7c60f79ebc6203cedc5989dc5c917af823","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-11 09:31:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c48dcd79f8ffb8226b4ed59290546ed4a146730","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 09:31:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d8dba0ba9092c80cee707bf98e687897eb260ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 12:03:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8f3ccac71fc4a2dbeff87dd7bfaceac : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15913025891f402cbb9f3ebe42717cd9 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45755a4aa5e44b3ebb206b3f44064e69 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46be0ddcf6434f54ac8ecddf44a5fb85 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae0e59dbec8e46738f38f2528705f27e : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d9386096c34ae684c1fc4a53035480 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/181d9fba48ae462dadcd46e607388c1c : SUCCESS in 8m 21s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/7d1b23401d194c8d890988bb762851bd : SUCCESS in 4m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/45e2390d4e1848afb7eae7c3ea2f426d : SUCCESS in 2h 26m 40s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/399444c73c1d425c9d93619766866ae0 : FAILURE in 2h 03m 41s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6e29cd17b1844866b32dd1f7e8072502 : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9afe4fee0f579ce2c2cab16cced2de637a1fbed5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-12 01:05:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a6dd3b138489364150a2ae0356d086b05dffe984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:28:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bdecc773e19a4963a947e0658adf81ca : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfb885bc95024d53b2d64cc1a1487358 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77361f16d5e24ed58d8ece6918ab81e9 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b0d8eecd0c43a897c3ce86e18e8d22 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c2c9a46ac04faca65e0098c6335c45 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91bdc6d7e67342fb95678cf0f5cd89e3 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c961c8fc01db433492b44c96a0e295cb : SUCCESS in 4m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6b87c9c641545c6a785fbe1acd525d8 : SUCCESS in 7m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c4673d89f9c4e109fbe35327fa074fe : SUCCESS in 5m 03s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/fa0190254e5d4c4f92a25903291a97d0 : SUCCESS in 4m 55s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/84622a9030b54f8ca2836a4a77786810 : SUCCESS in 52m 38s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8aaa0463e8544b08bfa466b25662bae3 : SUCCESS in 2h 16m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cd914a44bcc7484fbd8faaaa8f952ade : SUCCESS in 1h 42m 32s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fd676976b4ff4449869906aa9bdff160 : SUCCESS in 1h 34m 38s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/10a2150f389b4c8a964050233ed2f769 : SUCCESS in 1h 31m 24s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/397082c7de1f4b9bb48bbefd43310a37 : FAILURE in 43m 10s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/08ee4aa440c84e95b73a412e863bb544 : FAILURE in 42m 15s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/45c992b8a51140f99ce87ccd91b45356 : FAILURE in 1h 05m 50s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/1509cb607e3c42abb500d304b6101e0f : FAILURE in 1h 04m 00s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/87664b7db55e43dc8dc56ae877ee334a : FAILURE in 1h 14m 10s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a701ab9013284b8a9196eb6067ab9804 : FAILURE in 1h 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a85093f6c920632203bff8fa9232714c7a7bd6e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:28:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c8dc42b2170ca9e9762a6975eca9959dbf7629a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 05:52:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0621f2a81e24d42907c67dd3d337626 : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e559f37bd28249fdbf22facff072b80f : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d494859b67c44ee3855a256725803a59 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b5142702efa4418b806cdbc4d811693 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/801d63ce62f94921ad0cda71f8283335 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8554fb7703474393bba4279c3276b606 : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/224bed9d380c4572b872a08bd6875894 : SUCCESS in 7m 39s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/a542590f89b74b528a0be7eea0b25ad6 : SUCCESS in 5m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/bca50a46b6d94ec4b48543378a71a7c7 : SUCCESS in 2h 22m 39s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/679700cfb8404281ac73d5c4032b0aec : SUCCESS in 1h 33m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/93d72ea64838407b82180ca986e2ad35 : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":2},{"id":"40e99b7013b05d01d542f627ec5e9bc51e47a5aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 05:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"78c73102b932c6786504d8591fb3175a295fb28d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 05:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a56cb9d6fb9e4ac69b22174924190ce7 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315507b778674dbda466e48467e01aa7 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f16020392d9eb81727a97608f8e9be432f97c388","revisions":{"e1288345f90dd1248c2179706eaa6082a4d5d1b1":{"kind":"REWORK","_number":1,"created":"2019-12-10 11:07:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/18/698218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/698218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/698218/1"}}},"commit":{"parents":[{"commit":"0907c8ec61d5af07f2cde286ffe2828c82f0f836","subject":"Merge \"Fix delete instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0907c8ec61d5af07f2cde286ffe2828c82f0f836"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-10 11:07:11.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-10 11:07:11.000000000","tz":780},"subject":"Add quotas resource operation in API doc","message":"Add quotas resource operation in API doc\n\nChange-Id: I2108f2c6bf067ce228829de9c4e7e47005acf358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1288345f90dd1248c2179706eaa6082a4d5d1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1288345f90dd1248c2179706eaa6082a4d5d1b1"}]},"branch":"refs/heads/master"},"f16020392d9eb81727a97608f8e9be432f97c388":{"kind":"REWORK","_number":2,"created":"2019-12-10 20:07:28.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/18/698218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/698218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/698218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/698218/2"}}},"commit":{"parents":[{"commit":"0907c8ec61d5af07f2cde286ffe2828c82f0f836","subject":"Merge \"Fix delete instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0907c8ec61d5af07f2cde286ffe2828c82f0f836"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-10 11:07:11.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-10 20:07:12.000000000","tz":780},"subject":"Add quotas resource operation in API doc","message":"Add quotas resource operation in API doc\n\nChange-Id: I2108f2c6bf067ce228829de9c4e7e47005acf358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f16020392d9eb81727a97608f8e9be432f97c388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f16020392d9eb81727a97608f8e9be432f97c388"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
