)]}'
{"id":"openstack%2Ftrove~698886","triplet_id":"openstack%2Ftrove~master~I5abe9091ba203297dc87db5fba139179166321f7","project":"openstack/trove","branch":"master","topic":"health-check","hashtags":[],"change_id":"I5abe9091ba203297dc87db5fba139179166321f7","subject":"Support HEALTHY status for db instance","status":"MERGED","created":"2019-12-13 11:57:31.000000000","updated":"2019-12-14 16:55:45.000000000","submitted":"2019-12-14 16:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698886-1576342418844-349857f7","meta_rev_id":"7dc64816d8787071752805e1a7989886c52974fe","_number":698886,"virtual_id_number":698886,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-14 16:53:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-14 09:49:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-14 09:49:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 16:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"456a090e1a0daa7c4ffd92ed5ebe2aca0513e121","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-13 11:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7253b88d3bc7ebc47dbf718a52513b8c244c6566","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-13 11:58:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ac9d2dacaa5509105c2ab51bd48a63aff028d3cc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-13 11:59:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"133b2f8ab54d0e4d2fcbca676a0d0f90cadddb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 14:16:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4abd8764b8b8478ea297c9b3ecee20c8 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b344410be8b74342ae15a59f47f429a7 : FAILURE in 10m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff47769db47a406695d166964a43a393 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947f60219f78453d8318aaad5790dfed : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e8caf071821409581d802ac80714e31 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78d04134ccc64ebdabcbbf0f56e9d07c : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ffaa4b154149729cbc0c09512dc240 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd6de7bcac74161bf5e816588b8c755 : FAILURE in 5m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/504d58ecbab8405184310befd83815ba : SUCCESS in 9m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/663c34d66e47458c82f45efcd86efee4 : SUCCESS in 6m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/eb0320a74664424489be62b1716ddb0f : FAILURE in 1h 01m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/455d183f46194f84b3072ac528bf7574 : SUCCESS in 2h 15m 41s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/18ee3689d6e14e6e8e7611867427c584 : SUCCESS in 1h 49m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d30d5bec1cca4b338c081604c9eb1e5d : SUCCESS in 1h 38m 38s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/060a4f2b23614f06b16b7eb8f3d7baf5 : FAILURE in 1h 04m 19s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9a21106f3a23409eb3f2e9ca40441641 : FAILURE in 39m 30s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/72e29550b4b7495fb0504d31879d21c8 : FAILURE in 44m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/eb2b05b21e3c48d7add922960a2d874b : FAILURE in 1h 07m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b94207c07e8439a86946412e1960404 : FAILURE in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"96ca3cd03caf90c449cd4447c09d40a136fa93cd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 07:39:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c23b5cdb463b22fc9604289b46354914db716bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 09:46:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e4e19c11f144b39b8a8963654e7899b : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cf6a45b220740e39526cdbe0f604176 : FAILURE in 9m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2aec371bc494e77ae2e3b5433afa28a : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc0d31d032e4534a75f6ca8e8a79625 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8baa9b4718bd48c68f589b9f7d9ee00d : SUCCESS in 10m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/704a498203fc493a92f68d78efe39ef0 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47005ac62548449e9ae8c6c449b8b8a1 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01d6847de3694627a5a872b3e33f6cad : SUCCESS in 9m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6d0e75674554a37a84448b7a2e9fb3c : SUCCESS in 13m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c58d8ce86be44b4394da75ebffd4663c : SUCCESS in 7m 15s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/95e7d004befa43fc80b619ddb194ed8b : SUCCESS in 1h 25m 11s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/64178e2eab3e46779c7da0612613f0b5 : SUCCESS in 2h 05m 35s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/188861c5a9a44245bc143bf0b3b89317 : SUCCESS in 1h 35m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/077ab2ea6a8940bfb6ab6f921a618716 : SUCCESS in 1h 34m 52s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d586bd825db2471091f43add6481c30b : SUCCESS in 1h 32m 07s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8f3872c10e2c42ce816c6a3f5048f761 : FAILURE in 40m 49s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/031a752344f144b99a5d06e56149d5b2 : FAILURE in 48m 08s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ed7bd64d565344bcace14a9c791d5671 : FAILURE in 59m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/681a9cb3d7c44b56af804bbedb4bf5eb : FAILURE in 1h 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a5d7e5a480c2f77b461f8a700324b47839888beb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 09:49:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"619d8ec26ca2d2fd62b4066fe77a7b31e60652b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 09:50:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5f3add304e16db449a70019796fff00c512f921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 12:11:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8717db7011094bb39336db46b261a0d1 : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61e2df9d632a40058addec3d2bb37bed : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c94d5ddd8b542c7a0c4147ed6bdd00e : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6079074519344bc99bc0b548cf8ad70 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c65523b8158c4a76a504e21d63e8a1c0 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7527cb040e204f45ab4929b2f4b27672 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5fef463b85c4177960b9e6fa9e063b6 : SUCCESS in 7m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/017faab7124f4f55879bd83322e3264f : SUCCESS in 9m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3a5dc73ee0f640de99b2ca3121ca98d3 : SUCCESS in 2h 09m 52s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f7a7e9e2721745bfa2739df638dc171b : SUCCESS in 1h 51m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3f04316d2ed14a3583c448e3ff19f294 : FAILURE in 53m 24s","accounts_in_message":[],"_revision_number":4},{"id":"361a6ea46da4992519a495d4e28a816eb155d86d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 12:26:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"33aea5a36fbca72a7f30eee1039b412e13956955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 14:42:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/348d17246a1240be915e59b43b66ab99 : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0573bf5cd649d980b635167ef518e5 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64fb03b65992473c8379730523c0045a : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61736c1f8b254b8581170a54bff6de68 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6fe88f19c2447a834c96c0cdc0fb1a : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/309c9f19eda7420fb2d93a4862800114 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b547e7c870b42aa8377aba947a86780 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f0671e79e2d424b92631f82c6456062 : SUCCESS in 7m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3bd95ff5721c4b628731f99becc76dd3 : SUCCESS in 7m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c908a47b1ba408485b29de61af2c0db : SUCCESS in 5m 34s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/179eabd625bd4594a31caa5c50d58664 : SUCCESS in 45m 33s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3849111221074e40927f77a03f768f52 : SUCCESS in 2h 14m 41s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/989ab893c1dd47fd83d8e7c1e9ea2899 : SUCCESS in 1h 42m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3c41ddc6923a4fbbab75b25374fa9bd3 : SUCCESS in 1h 26m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/54eeef6e17254492a63d0328867f6794 : SUCCESS in 1h 27m 21s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/13d5b66001d14cb99447482b080e769a : FAILURE in 37m 43s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3b988207ab7d4a0a94259ee5179e973e : FAILURE in 43m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ec7ab98f5031478489ddbc7b52786c36 : FAILURE in 1h 04m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8bf87a73d4844b38342ffe4db0b1ebe : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d2e31d2b394fc205c2ee02977077c07db271f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 14:42:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f739b68f6a7e656a03c0d5c181fe14ae6690b460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 16:53:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6b95e67e12124379820cea50f17e682e : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f1093269214496a5349f77bc4d604d : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5613de60048b44238c18249d3d00817c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f79aecb2a97c43b5bf4ad82c01f8b611 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f893201b4854ae1a6b87abb6ad3273a : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e9de0557ad4086a2524830b1c16c0c : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a06f32f44314eab91ae75c9e91a7de1 : SUCCESS in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1663194425e24a4188da4f4f9ea95f0f : SUCCESS in 8m 23s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/908aa739f35d4c84918d9f7301b38be9 : SUCCESS in 2h 09m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0dcce502f3af482b9cc29cc5dae2fbc3 : SUCCESS in 1h 52m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6d6b2d3146ec4d868f8f513aef6227bb : SUCCESS in 1h 34m 32s","accounts_in_message":[],"_revision_number":4},{"id":"045103630a678119b4e9aae361702a5d6ffe60c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 16:53:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9cf3676e8084f9277d1cda3d8e0213f3f1ae2443","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 16:55:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f293f6e59e134a6ca873e554a3a2c7e6 : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ae6c0be1a04c0994e38ef4b6155907 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b60805606424e4e914293d6f376656a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a0a10f0b947c63ac06787b490afd0ebecef1477e","revisions":{"46db663b325861cfcdb94a7cc069ae945d962ac5":{"kind":"REWORK","_number":1,"created":"2019-12-13 11:57:31.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/86/698886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/698886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/698886/1"}}},"commit":{"parents":[{"commit":"0907c8ec61d5af07f2cde286ffe2828c82f0f836","subject":"Merge \"Fix delete instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0907c8ec61d5af07f2cde286ffe2828c82f0f836"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-11 11:26:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-13 11:49:49.000000000","tz":780},"subject":"Support HEALTHY status for db instance","message":"Support HEALTHY status for db instance\n\n- \u0027HEALTHY\u0027 means the db service is responsive, \u0027ACTIVE\u0027 means the db\n  service is alive.\n- Remove the CI job fakemodetests, but will add similar testing task in\n  the future.\n- Fix the periodic CI job\n- Remove MongoDB and  related jobs\n\nChange-Id: I5abe9091ba203297dc87db5fba139179166321f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46db663b325861cfcdb94a7cc069ae945d962ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46db663b325861cfcdb94a7cc069ae945d962ac5"}]},"branch":"refs/heads/master"},"24385aa23cb702396dbdbd2cec3631c061979c0a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-13 11:58:08.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/86/698886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/698886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/698886/2"}}},"commit":{"parents":[{"commit":"f16020392d9eb81727a97608f8e9be432f97c388","subject":"Add quotas resource operation in API doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f16020392d9eb81727a97608f8e9be432f97c388"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-11 11:26:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-13 11:57:54.000000000","tz":780},"subject":"Support HEALTHY status for db instance","message":"Support HEALTHY status for db instance\n\n- \u0027HEALTHY\u0027 means the db service is responsive, \u0027ACTIVE\u0027 means the db\n  service is alive.\n- Remove the CI job fakemodetests, but will add similar testing task in\n  the future.\n- Fix the periodic CI job\n- Remove MongoDB and  related jobs\n\nChange-Id: I5abe9091ba203297dc87db5fba139179166321f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24385aa23cb702396dbdbd2cec3631c061979c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24385aa23cb702396dbdbd2cec3631c061979c0a"}]},"branch":"refs/heads/master"},"cee41f6457f702bff8e409f6006525604528ad24":{"kind":"REWORK","_number":3,"created":"2019-12-13 11:59:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/86/698886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/698886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/698886/3"}}},"commit":{"parents":[{"commit":"f16020392d9eb81727a97608f8e9be432f97c388","subject":"Add quotas resource operation in API doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f16020392d9eb81727a97608f8e9be432f97c388"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-11 11:26:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-13 11:59:24.000000000","tz":780},"subject":"Support HEALTHY status for db instance","message":"Support HEALTHY status for db instance\n\n- \u0027HEALTHY\u0027 means the db service is responsive, \u0027ACTIVE\u0027 means the db\n  service is alive.\n- Remove the CI job fakemodetests, but will add similar testing task in\n  the future.\n- Fix the periodic CI job\n- Remove MongoDB and  related jobs\n\nChange-Id: I5abe9091ba203297dc87db5fba139179166321f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cee41f6457f702bff8e409f6006525604528ad24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cee41f6457f702bff8e409f6006525604528ad24"}]},"branch":"refs/heads/master"},"a0a10f0b947c63ac06787b490afd0ebecef1477e":{"kind":"REWORK","_number":4,"created":"2019-12-14 07:39:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/86/698886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/698886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/698886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/698886/4"}}},"commit":{"parents":[{"commit":"f16020392d9eb81727a97608f8e9be432f97c388","subject":"Add quotas resource operation in API doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f16020392d9eb81727a97608f8e9be432f97c388"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-11 11:26:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-13 23:55:56.000000000","tz":780},"subject":"Support HEALTHY status for db instance","message":"Support HEALTHY status for db instance\n\n- \u0027HEALTHY\u0027 means the db service is responsive, \u0027ACTIVE\u0027 means the db\n  service is alive.\n- Remove the CI job fakemodetests, but will add similar testing task in\n  the future.\n- Fix the periodic CI job\n- Remove MongoDB and  related jobs\n\nChange-Id: I5abe9091ba203297dc87db5fba139179166321f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
