)]}'
{"id":"openstack%2Ftrove~699061","triplet_id":"openstack%2Ftrove~master~Iabcfad81343a71304b843b3a7778486253220d20","project":"openstack/trove","branch":"master","topic":"updated-at","hashtags":[],"change_id":"Iabcfad81343a71304b843b3a7778486253220d20","subject":"Show service_status_updated in instance API","status":"MERGED","created":"2019-12-14 09:11:09.000000000","updated":"2019-12-19 16:02:49.000000000","submitted":"2019-12-19 16:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699061-1576771240305-716aec10","meta_rev_id":"0d57a8e2ffd28850e4adb6025f9141a4b64ef85f","_number":699061,"virtual_id_number":699061,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 16:00:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-18 22:13:46.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-18 22:13:51.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 16:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e446dbec2dfef53c899ab6a1f0c1346f895216b5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 09:11:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bffdeb6877078125a310b77cf3c4f8190b18b318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 11:09:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a053820d04e42268f7ea269d0fa04cb : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2544ab40136f4cd383f2935de8b6b447 : FAILURE in 8m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12cd4b7a52994ae589974a22979d9e84 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cae41b01649479f952fe4423f6a2609 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec28f3b412b44b48c43742b3a87f3e0 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/931e10b1a25b4f2a91d3997587693862 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3351a1fbeb24aa09d85e1dd9406b7bf : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39575980c3e449e19087179137e9f554 : SUCCESS in 6m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7aa1d6f65b3432a8b2a0e7b83b4d244 : SUCCESS in 7m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97554566e86e49ec8a01d72848f053b1 : SUCCESS in 5m 47s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/34d34313a8e04e59a8cefbfd7e362060 : SUCCESS in 52m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9ee1d9784c9d4c9e982e3f7ca62a93e2 : FAILURE in 54m 10s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5fdb3c31e50f4dc18cfec63208b84d8e : SUCCESS in 1h 29m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0dc18501fed94de5a24cf0727083db0f : SUCCESS in 1h 34m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/775a35b29fe34c5c9b1e10978039688c : SUCCESS in 1h 28m 59s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3ae3313648454856a758d3be61a2362a : FAILURE in 33m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3fd0dc65c49a48c79b8a8087bcf54ad8 : FAILURE in 39m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/427e4361f72a43a89b43000f6bdd5adf : FAILURE in 1h 03m 53s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1dd93e31e9b409d91342279ee9a63df : SUCCESS in 55m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0550b42a0a7294062dfe2326d76b89e47210adb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 12:27:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa15b350d905d7355c0976236703cf040926462b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-14 12:27:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63260ea131f1e40a914c67cb39f0138380235f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 14:15:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/187d31a1cbbd4a6596e3cc8e6507ad10 : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47d3ef7f12ca47159d7ae0a1734b2f66 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b62689d2b9e4332b72e6692901cefdb : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e1ab9093a84eabbf8d489c4e6c7074 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdbc43fab0b743f9b864cace80087479 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb3f949ff7ae4552a9d73ea9806478ee : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ade23674a2429fb01e2daaa08c6276 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/febb87f4b54d4d4b826b6491cf2ee1e8 : SUCCESS in 9m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1647025104304980b5d1c208f4c87873 : SUCCESS in 12m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dcb0c349da8749bda4df6bf777ca29c7 : SUCCESS in 7m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2188eb6d63dd4a5f8c9ffc304401f7e6 : SUCCESS in 47m 25s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c0e23c7eeb2e421f9871792671a93bfa : FAILURE in 55m 05s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/169f9fe3cbd54ff2b6895725867d7a33 : SUCCESS in 1h 33m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4074d5121058405287d537627796b30f : SUCCESS in 1h 43m 05s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ea3f579d0dab4abf8818032a756683e6 : SUCCESS in 1h 21m 28s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ebc36df492274c3ca1443784a9502f75 : FAILURE in 42m 00s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/61c8631713f147c197e1c57671356c2c : FAILURE in 39m 36s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/412b9f6e95924dac85c1da16e7e29919 : FAILURE in 1h 08m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6c8f619009f4081b9f289b9de29970f : FAILURE in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e14eeed969d6d5f8f45d9bae18966e2e35a7a55c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 01:20:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5005eaf729e1608605855c45d78b6cda815c090d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 01:59:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/681f6237a1444410a7bfddb91cc9f250 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d0542ce35c042d08fdf9971e77d0ecf : FAILURE in 7m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abb10b1188e940629a3c298270b96351 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5e96adf96b4035b42590f154864e5f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a393e8c04c842b18996d5c95d27d71e : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2661824a8b5249cca1ec6382564bc2e9 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5994db7481a4dc4bbae16af5755fda4 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1494d7c6dce0424ebe46356d331753da : SUCCESS in 6m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e31d390f1dcc4db88b8d07d0c3d92737 : SUCCESS in 8m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/274547266aa740aeafec69aa6a24b72d : SUCCESS in 5m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/17d58090dd854abca3e2ca0aebe2c066 : FAILURE in 11m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8de21bac30e340b18651e7b63b37f8f8 : FAILURE in 10m 27s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5af19ce0c337420ba15ce5934fc8b103 : FAILURE in 11m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f6ef9a407ad44a979705fc198cf4a540 : FAILURE in 10m 45s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c9559d0e3a0b4cfd93a2dc48fbae32da : FAILURE in 10m 48s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/160f88c69f524007a551cff375f2b802 : FAILURE in 10m 31s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/70557f3bce4b491bbb2a15f41a1534b9 : FAILURE in 11m 00s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/bb60ab83a32e40848ce880c05fad093d : RETRY_LIMIT in 8m 03s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b92fdb8b1824a6a902e3342e538479c : FAILURE in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"790640f866b1813655590080aa601dc15e23cc6d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 04:31:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6579dd48709c37e0b847806646faaf5f153dc8a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 04:51:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/973772e355074a50ad8239d01fdb0c18 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e834d1f2bc6410490820daadc517b5e : FAILURE in 6m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db4b7ccd0d8e4f258e5f5307803bdc62 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b856c4429e4a259edfedb0eb3b1e2d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0aa47713e44353b4503efaa06260c1 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab5272dea8224d548efc056f0cde07c5 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812f21e1a05c476583076e5c872a4b15 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feee269bdc254e5fa49b9a3b4b5ee964 : SUCCESS in 6m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/720b588dffba4e93a75504fcba2bef5b : SUCCESS in 7m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9bc45ead6e041a182b97c6f3860d903 : SUCCESS in 5m 15s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/095c7ff045de44e896ebe216522e559e : FAILURE in 13m 43s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4ef3f75379f445159ae5cb794ba96577 : FAILURE in 12m 58s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/32735ba5f95342ffbb26129f0bfb8f10 : FAILURE in 12m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/840f95cb82b84ced9b96cd2ad0c5bd86 : FAILURE in 13m 49s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bdfe95e68df445d2ac03917a5e3ba3d2 : FAILURE in 13m 33s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/12f8002f2b7d44b0bb42e636eb69a18b : FAILURE in 13m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/657b6eb99fb14bbdb618df62f2edd7c3 : FAILURE in 9m 55s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/bb91012b43574d558c128cee7da7634e : FAILURE in 13m 04s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08b92315c44e4725aa36e637c7f70178 : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e03cb775cb3fdce22122b91a89ab30e817694605","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 09:30:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"81933531f3bb866435b16ee6b3496f74f732faac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:52:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce4d3730f43a48d5ba929d53f7c88cad : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d64a52712dac4b308b2eb5a2fa399efb : FAILURE in 6m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8775b573a3b4ba3b4c5f8ed7d33ee33 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b9d4a9ce8c46248208d8303a9fc653 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca81de2b2eb4b51bf44c688926e6d2d : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77b447c22957474a89af7aa775aef02f : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ed54e978204646808c2694d9339617 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40a975a72a214da98480c62ca99c88a6 : SUCCESS in 6m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a294ed9b0ca144858c0f05311b851bb1 : SUCCESS in 7m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/526dbf5e5f8047bda0e7f8ebb3ee5841 : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/64ea78449f5440068b2d2dcd49e218a4 : FAILURE in 11m 03s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/32c9fd74f67b417f96e039022daa02cc : FAILURE in 10m 28s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2886e23f1986435e98b1ab0c9cdf61bb : FAILURE in 10m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/154686027cc8462f9bcaee6da8241505 : FAILURE in 9m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/76e6b04407bb4d64bd4ae9dd3c854f0d : FAILURE in 11m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/64f8d64368184ae7910b9ce0740b06b6 : FAILURE in 9m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/52bf91d22ebf417b9fbe8ee5c3ad24f2 : FAILURE in 11m 19s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b7fe39e01dc343c486522af01b229e2f : FAILURE in 10m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/621eeaaa14d34555a88e5abceb3acc5b : FAILURE in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85e8cbcd3d6e79175bad0f5d5d336ee3cead4624","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 20:48:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5c5c03f85c3b9128acab9b68f65263b1676a67d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 23:14:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3864d0257f624c269cd684d40173bd97 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17343e82bdf24058820cf91416cfb7ec : FAILURE in 5m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c72a9163c042acafbc8bc479c18ca6 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd7d9711f454b6f9dae681f22df087d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7a014b76f0645e9ba49728fd791afee : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c7489f820834e8fb5ec0a8bcc7fa54c : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c15c6bf74eb42cc98e79a4a101b0ff1 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae9d3057c87468684e287293c7794cf : SUCCESS in 7m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59106d4208254c688ad75000739f7e09 : SUCCESS in 7m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98d17c91d11e4f91ac8e76043d34ff7d : SUCCESS in 4m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1d46d5d5479d4ec586d838c6b1489032 : SUCCESS in 53m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/61deb32b0b8745bbae23f43bf3887d6c : SUCCESS in 2h 20m 29s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d16a6c2a2d91433e81c488129bdec16d : SUCCESS in 1h 44m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/19ac4783edde42c2af20d8d279fea42d : SUCCESS in 1h 47m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a8d77c56bd5a4fee9504cfc70a1d6ace : FAILURE in 1h 11m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c6a8b4a63b084157992dbd237e663791 : FAILURE in 40m 04s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/724af9cf851d4ea089e3e511de7c96b1 : FAILURE in 37m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/9219536fac8f4cbeb8b865edcc10fd44 : FAILURE in 1h 07m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0f958bb3db0480d8128455ac75f614d : SUCCESS in 53m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae3a5bd7ce217f936a086a37a87b79709e15a9c2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 00:45:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1a98b648ed5a00f7a0d892f3f0eb6d768e7e1666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:35:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7310f29447284c669ccb1c805f24fab8 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c6b918a3d4546deb9adacaaf59f293f : FAILURE in 7m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46e4c28c11b448b2a616f9a338e975f3 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b1acddcc9142dbb4a1f922a0ce8402 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4970a35795e4506a420f884b9e99da0 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf44863036f94f6586eefa3847a7688b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0898e1aaf0384bfa9da11f5c8702adc6 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd3e38f72c864fd89e982e1c8bcd4122 : SUCCESS in 6m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fb1798bf6d641a19303b39bcb28c39c : SUCCESS in 8m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08a90554586644ae8610767c10af218b : SUCCESS in 5m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2bfa63f841f14196b01de183322e6b40 : SUCCESS in 48m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ce02d21de4ba43c1b9abcf5be651290f : SUCCESS in 2h 25m 11s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/99f0d9679e8b4a7bbc0d8ab50bf84bc2 : FAILURE in 2h 09m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/81a8c08e017d4de2b835278973ab18b2 : SUCCESS in 1h 39m 49s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/32de28e7808748caad7563155e97e908 : SUCCESS in 1h 32m 02s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a5b35e3ba45c4d7cbea2ef58776ffe05 : FAILURE in 38m 16s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/65bf420f7d1e434c8012424ea22475a9 : FAILURE in 39m 19s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0e0d25cf3ea1415aa09c3f92f180979a : FAILURE in 1h 05m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/523f15b41e2c4dbd843f088822bcf0b6 : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8fc1cb4016278a88d2255d51ad69059e1684de2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 03:37:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8774e2dcaefc4b24962b6b2a64f6f0f37de4f201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 05:37:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db53cd2f9db148978aaaa7daf7231a95 : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/387f46b48e774561b9cbb17674af1539 : FAILURE in 7m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c9aa1f7c4f3465caa09bf7722d91d52 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91acf5b66124c5bb46c2774670be8d5 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43941f7cd7044152abd0bd1978d02434 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d39967b416a4e8da946b784b04c694c : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c6421cb237465b9cfda9cc562f01ac : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93e676bea6974c8bab05aba0909bc556 : SUCCESS in 6m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0c1c9e1f1bf405e9352115a7a65c399 : SUCCESS in 8m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8bd605fef40e437f810efd4176b00f13 : SUCCESS in 5m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/97a37dd19bf24e43b05b955502be4c00 : SUCCESS in 53m 28s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c95efd3cc384490080d282d3ae4a3b16 : SUCCESS in 1h 58m 15s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6fe73b3af75f40a4a01425afc53a6f7c : FAILURE in 1h 37m 02s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e4355e5c7ff74cdcaa9615964a8be3c9 : SUCCESS in 1h 48m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/1fe89219d7444927a45838d7a8c0d304 : FAILURE in 1h 39m 20s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d0cc6cbb95224f3ba4e9046c6e35dd1d : FAILURE in 44m 28s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fa4b780e7595457abe3df569b355adfb : FAILURE in 40m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/74d9a952cc884a99831a09542ddb5a83 : FAILURE in 1h 10m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a44786783e498fa762a0b8ad5fb063 : SUCCESS in 51m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3a12cc5ade753c2ec509ee8c32a316462263a65","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 06:14:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"849b7ffe45e79cd38140e75d7b425b5054a3f007","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 08:35:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8b07ad6102a1c30de45290eb590f05d1e73fa446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:18:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bae8075f99540d2adc515530a20f369 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c4cd73343f411d80194fc101449893 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/644b5c603aea4d61bbceaa625ebf2ca3 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68fa7ff9103d4f33abd67fcf193e00b2 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c01e48c900a41c5a2245f8aa7c6d64b : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef762c56a4514a5fb370421d073b92b9 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bf461a7b9f40be964b576bdec5f3a2 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf5a35be2ce479d806b09165048b5fc : SUCCESS in 8m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7a84ced559044fe9319ee3fe58a4f1f : SUCCESS in 8m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3cd90177a2fa4e85a02a27380287f5c9 : SUCCESS in 6m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6fbbcc85c08e4d2dabc5f0b155597067 : SUCCESS in 57m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/808a49b6d0974a68a112b73bdc10a564 : SUCCESS in 2h 36m 36s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/872704bdc5474aca93be6899c5d9ea80 : FAILURE in 1h 38m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/526f8a0aefb04950b687db472bd1d2d4 : SUCCESS in 1h 55m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/043d97dd45ef482bb6c251ec531b5a27 : FAILURE in 1h 37m 10s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/71550a54c7f54cef83fab740c9290737 : FAILURE in 49m 34s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/903d636065e5419dbc24f3cf7cbb2276 : FAILURE in 50m 49s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7e06f7c5fc3d482eb33693cfb7872f73 : FAILURE in 1h 09m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e069be14002a45eaaa1ae58524cf629e : SUCCESS in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e823e6aa2070fb35a1d02ab3731f8d1a336fb92","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 10:47:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"18917f0d1d728725478efc924408064838f39737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 13:09:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/345120334a494578822be59ec3931544 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba4f9dfc1c704321b27be0e2ea46473d : FAILURE in 12m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b2ba012cc44472e946db2b2e994c973 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2b76be35ee42c585b332ca5906e705 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af3363ca09f647c2b9502f56688ca5eb : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/652e6bc0548342a59eb886bfb7c22405 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b757ce5eca9540dc8b2dd401b0b2bc12 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/056450a65715438ea90374c6bce9e300 : SUCCESS in 6m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be16cb222bcf43229aee6a7cf526d7a0 : SUCCESS in 10m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d318b8cc1c748e78557c1bd0b7959ba : SUCCESS in 7m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/092e892c1ba94c33ab7d770e3f2973d7 : FAILURE in 59m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3455309195c6449e821d4698525519f6 : SUCCESS in 2h 18m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1d7d63423c744925a0146fb278ff3fbd : SUCCESS in 1h 37m 56s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/bfd3eba48f584178b008d8ebf99f6c8b : SUCCESS in 1h 35m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6acafc0cf38041e8a83c30853e774e09 : SUCCESS in 1h 36m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/0144a0f9a30b4070bd05ab6232ae7403 : FAILURE in 38m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e5bf10a2b4844bdf949961d740a7cc13 : FAILURE in 40m 05s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4c2b05eb1f4c4e129bdd9353118e1c11 : FAILURE in 1h 10m 08s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5870d8d245aa4c30ba48ce6fd74e1cca : FAILURE in 1h 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"799ada9adcabfd8be14c22ac8fcb5d47f000b668","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 18:42:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ccfef03ccc0a314a0550c96f927780da1bd9d1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:34:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de5bb295c3384214b73489ba7fbaf713 : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f979b2247ba49b7ad03cd1dce3aef43 : FAILURE in 19m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe2b41cb5cb4059ba68b22f1c41e601 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef3647a6d9f4daea09d8230e89cf8f1 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f885e8b6419149d087b8954915fb86a1 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76ffeb74d21240f295b5463421da3e0b : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d6686cfcc94b488f2b2ce7dfe9f962 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c95337eb9ef456abace3b847e94274c : SUCCESS in 8m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a89fca7fa7849968d64d35286b5e9d8 : SUCCESS in 9m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77a78645c26b4160954f5bad72f86750 : SUCCESS in 9m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a9762a02b7b94e8ea9015e144b2ce58a : SUCCESS in 1h 11m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/61e96e1db35d4de6bf6a8ca96e413327 : SUCCESS in 2h 43m 24s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f1f1621e1475434880613ea43ef70504 : SUCCESS in 2h 10m 31s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/dd9e100c957a4fe980f5b11b8bcfcf9b : SUCCESS in 1h 57m 12s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2c13fe75e30c48dbb88106e05cf6a19c : SUCCESS in 2h 06m 00s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d0ebebbc5a924dc09a1210fa9efffd26 : FAILURE in 1h 04m 04s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ee156ebd88744e67b819bb9b1e1aa8e3 : FAILURE in 1h 02m 50s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7f15aae3139940ddb0b0c7883f35a1f4 : FAILURE in 1h 33m 05s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa7d3444758e4affafc4102d6e827c6a : SUCCESS in 1h 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1e882dbed0d48f311b83ba38483931361307bbfd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 18:21:48.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8d8669e105c17cc5feb3961757a7ddd006e1f3e1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 22:13:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a0f047e8067bbaee56f5387e4384230737858c5c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 22:13:51.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"799eeacca94a932133f9014b164d4c4db56075f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 01:26:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b24bcae71bc4501a2ba8c2364ac9be1 : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3651bb6ae07140cc8a78c12ad5066573 : FAILURE in 7m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cade42711d3940f19ca352ed636a8c66 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa5bb2ad0ed44f88b011c0d92904d8f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc7171a88aba4b54a8ae9b7ee2573ee9 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/583cf45f175b4dd09066cad5f3f2076b : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b86d2efff8454796af830ef18fcc48 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e479b1f2e3434582b4d82c918144ac : SUCCESS in 7m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c528d98e8d748bfab581a32aa2c3fd1 : SUCCESS in 9m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37a0b6d677e2430d819cd87444d57aba : SUCCESS in 5m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/297ac2703dfb49b5be2c10bdfbf2652d : SUCCESS in 1h 08m 02s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/09a9cfe601c0468f816862fec790043a : SUCCESS in 2h 48m 05s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b4feaec36b6b48faa3606ee75dc3d37d : SUCCESS in 1h 53m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/80d21c3cff2d44b8ba1042970a23f4da : SUCCESS in 1h 44m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3f2687aed7b94354bf8544fd9aedb8f5 : SUCCESS in 1h 37m 22s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5904c8cd87a54713ab3ca5373f1e1767 : FAILURE in 43m 53s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e36f53ff4b9241019553f7672a76a3b5 : FAILURE in 43m 52s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0798af40af204ed4a7850388713326cb : FAILURE in 1h 14m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b93005ba948f4622bae4d2e1b9efe80e : SUCCESS in 59m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a32f824cc6580e777c74edfed78fe23ff4c47050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 01:26:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"88fa948ced150e00b751ab672099cea452823741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 04:06:09.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/895bf13de4a04cc7b5caa46bb06e7f01 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74178cc59ba2434a98c7b60c97e7c125 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4834f5893045f3919d9c0e23f81755 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a66d1b3b602a4414b1d06b3abe094a93 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24055df1eb734d298edc73dbbe332581 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae3aaa941294c4bad1af41cb3082577 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7335673f56d4efa88c93a8927463aa7 : SUCCESS in 6m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3bd9d2932d4408b95e6009e910b52d0 : SUCCESS in 8m 36s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ebfd6e5ec6364332a65e83a4a05ea53d : SUCCESS in 2h 33m 37s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/85c230a08fd64f99b5aee3c96acce218 : FAILURE in 1h 55m 26s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ce80a91e2b244ca4a5384a50f53ba9d2 : SUCCESS in 1h 44m 56s","accounts_in_message":[],"_revision_number":5},{"id":"ffdd13c0f745cdf6735c3d7409e330559111a326","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 08:32:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"323c994e43f1b2e50706e28d91e5857628456c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:30:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf20d98acc714bd7abc853552f687c54 : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b45e674b9d144eab6c52ccc63cded81 : FAILURE in 11m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/886fbf77a28847898d7a51e74c7741e1 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b3e602767e4347a93ba80973ea9fc4 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbfe3375f4fe4ac5876fe8fe4ecc1f8e : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89a9abff8ccd474f8a4e18b55e483b94 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456ce4d5888248c590ccc6702253ecbe : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c250c327ad5d45b786777a2834405a19 : SUCCESS in 9m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e9f82c57ffe4d048eb4396c975b08a5 : SUCCESS in 11m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee52fb8779e54eed86c719141ab9c068 : SUCCESS in 8m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/95851e44652a4861896548e05ff52537 : FAILURE in 1h 00m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dd0ce589d6054458aaad050fb234f341 : FAILURE in 1h 31m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f26f507a783c4afe923a79332661bbe4 : SUCCESS in 1h 50m 11s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f27d3590aafe471f87bbb940c734cf85 : FAILURE in 1h 00m 26s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a74bff2c6a754b138a59d3c45e036e1a : FAILURE in 59m 07s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e941ee2964464b45a258592db637f4a3 : FAILURE in 1h 09m 32s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3f4e13d1aa094cb6b0d1485e8856dd86 : FAILURE in 56m 18s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/215801d98c4c474fb2cd0a7f55bad1bc : FAILURE in 1h 15m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d8af216de2949e88615bab998c67a6b : FAILURE in 1h 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3b0c055aaa3842477e0cf9d0053324319bc728fe","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 10:37:20.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"476f25042b236a6fd6970e5886adc97b97447c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 13:19:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d332b228d7e4e6f93cccfae70e6d7be : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44aae2b7ce79429ca3bfad0af642a1ed : FAILURE in 8m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35fa6630e4f1444c91f82c15844b9a78 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a321f4a6014527bfc0759afe4f34e4 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fa378e1f272456caabb3fd7806e228e : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7aa2550010d44b9be3d8067966a9eb5 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49dc4b7f52f4d4ebf57a17a2e8549b0 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/299d1e0c7f11402d9614dbc9534d45c9 : SUCCESS in 7m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7de396e2c53d434586eebfa2b4045ef0 : SUCCESS in 9m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d106300521bb4a699a6cb1103aa7c595 : SUCCESS in 7m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/543afe5c1cfd4c4eb5f8af8d96f54ae4 : SUCCESS in 55m 00s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/26aefafe95ae4bfbbeb30258f39ca553 : SUCCESS in 2h 35m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/08e9d1c0a67840939510ac63390bc881 : SUCCESS in 1h 50m 16s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/48529034172e4490a48744890eab4c4d : SUCCESS in 1h 37m 39s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bf1c9a9efad640b681390552e96eb18f : SUCCESS in 1h 41m 46s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ed86210163dd45899b34637428795729 : FAILURE in 47m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/b0b073ca2fd8480b9a912ee382a8c064 : FAILURE in 45m 55s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4acf68f6309741cba2e1f848a8108c86 : FAILURE in 1h 11m 04s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6be3f132842347438a67aa07787cbed3 : SUCCESS in 1h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b1ede89f452b91cd01dac1f62b626caf533c0f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 13:19:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7ddf0fa2248a7244c32d0844d88c23861564553c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 16:00:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9631a0badee6b1077a7eb8cf2bd1b4a1b2b92b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 16:00:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4518c65b08104eed949bf2a19bfe336a : SUCCESS in 20m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5525b4e65ee740f8bf5b94b181377e1c : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e491bcb516e4142b77f13a0d3ab3c39 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f102a193d73455d98776342bf3e190e : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6132863908849e0b5a5b097791b9bd6 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cf9f18233b4eab95b52d81a4f47db8 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b584f597f754cb88869380d2bc320d0 : SUCCESS in 7m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c452c4ee5c646deb12faa14858cc22c : SUCCESS in 8m 35s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7a75604d5fcc44c2bdb754aebda91e3c : SUCCESS in 2h 35m 15s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c7349a5fc0254dc180624bf4a53eb47c : SUCCESS in 1h 41m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9441d4fc20434e49ba6f2429de731c42 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":5},{"id":"4c017c21567bb0c79a51894ee9ef1d52b5ec4ed5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 16:02:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ea692d692c9e4ba9b903d1c58c4cd6c9 : SUCCESS in 1m 19s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e847cf1b904af6a3738d3d778d0462 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2337e37f85c465fa38e26025c2be965 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"75b9b686b1c4fa2be1be00ef390035da01e331eb","revisions":{"15eb9944962b19d11e719a4f1983c8eea8a5f9e0":{"kind":"REWORK","_number":1,"created":"2019-12-14 09:11:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/699061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/699061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/699061/1"}}},"commit":{"parents":[{"commit":"a0a10f0b947c63ac06787b490afd0ebecef1477e","subject":"Support HEALTHY status for db instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 08:25:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 09:10:04.000000000","tz":780},"subject":"Show service_status_updated in instance API","message":"Show service_status_updated in instance API\n\nA new field named ``service_status_updated`` is added to the instance\nAPI response which e.g. could be used to validate if the instance\n\u0027HEALTHY\u0027 status is stale or not.\n\nChange-Id: Iabcfad81343a71304b843b3a7778486253220d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/15eb9944962b19d11e719a4f1983c8eea8a5f9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/15eb9944962b19d11e719a4f1983c8eea8a5f9e0"}]},"branch":"refs/heads/master"},"a5159c5b31fa0992f1486b8f6bae653db48b31f7":{"kind":"REWORK","_number":2,"created":"2019-12-15 01:20:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/699061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/699061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/699061/2"}}},"commit":{"parents":[{"commit":"a0a10f0b947c63ac06787b490afd0ebecef1477e","subject":"Support HEALTHY status for db instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 08:25:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 01:00:11.000000000","tz":780},"subject":"Show service_status_updated in instance API","message":"Show service_status_updated in instance API\n\nA new field named ``service_status_updated`` is added to the instance\nAPI response which e.g. could be used to validate if the instance\n\u0027HEALTHY\u0027 status is stale or not.\n\nChange-Id: Iabcfad81343a71304b843b3a7778486253220d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5159c5b31fa0992f1486b8f6bae653db48b31f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5159c5b31fa0992f1486b8f6bae653db48b31f7"}]},"branch":"refs/heads/master"},"5c4ad25aa92f861c975f8d9f2ff83f2b61e01a42":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-16 08:35:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/699061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/699061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/699061/3"}}},"commit":{"parents":[{"commit":"e048feb145d99d4584e0a4ed63fbbcd557ddf948","subject":"Remove some unrelated code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e048feb145d99d4584e0a4ed63fbbcd557ddf948"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 08:25:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 08:35:13.000000000","tz":0},"subject":"Show service_status_updated in instance API","message":"Show service_status_updated in instance API\n\nA new field named ``service_status_updated`` is added to the instance\nAPI response which e.g. could be used to validate if the instance\n\u0027HEALTHY\u0027 status is stale or not.\n\nChange-Id: Iabcfad81343a71304b843b3a7778486253220d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c4ad25aa92f861c975f8d9f2ff83f2b61e01a42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c4ad25aa92f861c975f8d9f2ff83f2b61e01a42"}]},"branch":"refs/heads/master"},"1e3f36af32d12140fd204d0895320645f9d6ad5e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-17 10:47:52.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/699061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/699061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/699061/4"}}},"commit":{"parents":[{"commit":"ec8a754198ac1bbb6954d559ea1d5d679c2de457","subject":"Fix the log related tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8a754198ac1bbb6954d559ea1d5d679c2de457"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 08:25:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-17 10:47:40.000000000","tz":780},"subject":"Show service_status_updated in instance API","message":"Show service_status_updated in instance API\n\nA new field named ``service_status_updated`` is added to the instance\nAPI response which e.g. could be used to validate if the instance\n\u0027HEALTHY\u0027 status is stale or not.\n\nChange-Id: Iabcfad81343a71304b843b3a7778486253220d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e3f36af32d12140fd204d0895320645f9d6ad5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e3f36af32d12140fd204d0895320645f9d6ad5e"}]},"branch":"refs/heads/master"},"75b9b686b1c4fa2be1be00ef390035da01e331eb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-18 22:13:46.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/61/699061/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/699061/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/699061/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/699061/5"}}},"commit":{"parents":[{"commit":"5472d8a57c7c9e82cca122396fea75b8ef05a64a","subject":"Fix the log related tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5472d8a57c7c9e82cca122396fea75b8ef05a64a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-14 08:25:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-18 22:13:46.000000000","tz":0},"subject":"Show service_status_updated in instance API","message":"Show service_status_updated in instance API\n\nA new field named ``service_status_updated`` is added to the instance\nAPI response which e.g. could be used to validate if the instance\n\u0027HEALTHY\u0027 status is stale or not.\n\nChange-Id: Iabcfad81343a71304b843b3a7778486253220d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75b9b686b1c4fa2be1be00ef390035da01e331eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75b9b686b1c4fa2be1be00ef390035da01e331eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
