)]}'
{"id":"openstack%2Ftrove~699100","triplet_id":"openstack%2Ftrove~master~Ic1d4982a479c2154d7ddcc807d863e2a37fea506","project":"openstack/trove","branch":"master","topic":"fix-pgsql-image","hashtags":[],"change_id":"Ic1d4982a479c2154d7ddcc807d863e2a37fea506","subject":"Support PostgreSQL 12","status":"MERGED","created":"2019-12-15 06:26:09.000000000","updated":"2019-12-19 13:43:15.000000000","submitted":"2019-12-19 13:41:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699100-1576762874787-716aec10","meta_rev_id":"cf16ced119d94e9e2eab2633c42512043fbc2b1c","_number":699100,"virtual_id_number":699100,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 13:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-18 22:13:32.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-18 22:13:38.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 13:41:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e4dc98bf756ec7e1c08239ba5ae2916908753a7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 06:26:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"973357d0b0ed5d3e9bbad2d36214964fe83e7117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:13:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/376311894e574cf58f2005e0a8c7c852 : FAILURE in 8m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c4134df4614f5b969153b96d6dbc95 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de41c747aa44a7a9fc4597ec6fdd7e6 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810a3aaf6d4b4058b9d81a663cc47286 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24fa58480caa4a34a50e56eba54fc49a : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb13abf0a0244b4bdfad752dfede07d : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a7b9dfe8f844e218d17b43b9c8c1576 : SUCCESS in 10m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a96da1d9d2c243d5b9e7eb5831c5fec3 : SUCCESS in 6m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ecbfd70dd4984cecbf63371ec6ee0ed3 : FAILURE in 10m 48s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1a77d04b11e44185a790fb9b457845a1 : SUCCESS in 2h 16m 16s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2f19096b425348f0a11caf76b5e7ebed : FAILURE in 14m 27s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/55e5827d8e294927aaed81dd8b8c4ebf : FAILURE in 12m 42s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3f926d6c5fe44531ba0ce94a61c9f2e6 : SUCCESS in 1h 45m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/b1b5357b3575447098ba6b1852d5e6a0 : FAILURE in 1h 48m 07s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7cacfab0421945b08572935b5c2836e3 : FAILURE in 1h 11m 55s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3530927ed52243dfae926082169847a0 : FAILURE in 1h 24m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcf44d3ec94f47809aa208abdac8bc49 : FAILURE in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e20871d376d0c7175c3c045e81602f10dec1a3b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 09:56:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ad6d1f83f0e184e0a754fa7f39afbc25a3cab88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 12:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8494166397d845c59f817381be1f7476 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38ece5d554ef4423a36a9f9dc9769cc2 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07aab65fe0f4035a581bcce7af493a3 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d01039022f9c425f8a67ab1d94873b71 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d78cb2e60174c95a132e936f5241034 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fe31098fd64ec4bfab70fe5fafa790 : SUCCESS in 5m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b022929f3ee457b9b543c76366263b1 : SUCCESS in 8m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf1d18f7ec1842d0980314a0d50fbc24 : SUCCESS in 5m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/49218998c1b14888924846446a7ac3dd : SUCCESS in 53m 30s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dc53774c32f84d3085aecf7614cfd313 : SUCCESS in 2h 24m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6197dfd8871a4c169c77d5ad647c093a : SUCCESS in 1h 44m 17s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c4ee795c2d634df18da761d8a2164553 : SUCCESS in 1h 38m 34s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a486ac0ca4f94388aef15b27a73f66f8 : SUCCESS in 1h 33m 34s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8f0771cd65d043558371dde6c40df322 : FAILURE in 1h 35m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/b987825c378b4108af9e378d02c33228 : FAILURE in 59m 56s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3394b6b2d25841bc87f32c98fd0c8560 : FAILURE in 1h 06m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e7c16afb4d5465d89703cef278cfc61 : SUCCESS in 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a85c5bea4b7baaf2e8e986d5315aa39a18e888e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-15 20:47:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAll postgresql related jobs have successfully created the images","accounts_in_message":[],"_revision_number":1},{"id":"2327d146b95f7485652519506fdf8833a49a39c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 20:47:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d751d028626b9d747db2f389ea2b0ca3b6350a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 22:48:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bf9f466a4de4091b0529164ee32a456 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff56386a20b94bbea7dc7d38191b062a : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74dd25cfa8e46c8a57982f2e5e443f8 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c16a8155ec1450dbd5100df7dbcf1f7 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d572b7d3e24fa7b5c2c87dc0d5eceb : SUCCESS in 5m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0164c2a413404ef496778fa0cedbc68a : SUCCESS in 7m 44s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8fb18f8f1cbb4969920989e9b21d3cd1 : SUCCESS in 2h 00m 15s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b871c60c92214777bd4705c0f175b594 : FAILURE in 51m 27s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/56c6d75ab46b4b638f8de6f439dd943d : FAILURE in 48m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e9dd030caaff31013baa946f13753db94c408a73","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 00:42:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ceb80f33deeb5d4a1f4163f0b0db504aae67584f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:22:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ad7406ae504e57bcb52e8ace9a78dc : FAILURE in 10m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66ff785f1daa4c28b57941d0b8cf50e6 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ffbb29c7a9445eaa4328ad52935062 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7941e5f1bc434e850207634cdf3495 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55f4bb671f27417b81b63b4c7691ad15 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d06ba750c29440a9e6dbf3bf2c26eb9 : SUCCESS in 5m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09f07effa50842ba81dd0c213e42e9b2 : SUCCESS in 7m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f031b29c3b9c42d3a312eafd635b78e6 : SUCCESS in 4m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5eda9df45d584f9a992e239beaad3bc2 : SUCCESS in 50m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4f684e3f2d0d48999e9cc0f92685b371 : SUCCESS in 2h 15m 33s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/91a6632d76524f1886ccc245753e2c14 : SUCCESS in 1h 44m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6d1238abe93a4664a516136ab363d2c7 : SUCCESS in 1h 34m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4ae9610c279749589d042eb1d85589f7 : SUCCESS in 1h 34m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/6a5d9669eff34383b9fe905a9b22bfa8 : FAILURE in 1h 29m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6160c338d59b4105bef62076228231bc : FAILURE in 57m 39s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ad5fc6f72bf2461e8b2849e4dc9848f7 : FAILURE in 1h 06m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64b8a61e12ee4ae1acb014674acf60b6 : SUCCESS in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"466eb3348a1315d97d2dea0e1d658c2005c43ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:22:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"848d719baf2bff8c9459666d61ff965b268493f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 05:51:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d2c064c390e4c96823e1da4b3dd8fc2 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8ef671234845cb8694c68b0b974625 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b6e8b5c66b49b39f8cb3f0efe9a190 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61da526f1990413d8ff73fbfc721d7b0 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6219c8202049899849737b6e516b07 : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22655009ff3f4615bd49bb237849c4c1 : SUCCESS in 7m 59s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/181fb093fc714d56ad2a0689983154d8 : SUCCESS in 2h 22m 55s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/191d2707760942f8b44ad74d5ae27c65 : FAILURE in 1h 45m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/68ee3bc9c2c843a1a0266e637faadf46 : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ad55b445930a91876c20081a124ee13227b868ad","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 06:15:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"092c1513a6754e9659bf9fd39754727e45f60210","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 08:34:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a84f6cdb81f0032526b6cb62c3f4083051db70fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:41:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a73641bd9d3b4a33855cc1b3eb20c7b0 : FAILURE in 10m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06fae65e600142ec8f4e8d150aac8b8b : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bb7a3571e54fd080dd8ae0da6c70bc : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f699be6515441bca7f2dc03b2815392 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ca31d7740124b7ebd57e6fa96d7034a : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c562ad570e904da19608b057d4fe9886 : SUCCESS in 6m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8c9a1af24fc4c65a60cc69603de3204 : SUCCESS in 10m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3514e6fbf69043669295c9f61fcec63f : SUCCESS in 6m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/286a114e546a4e38bd5cbafc5af21778 : SUCCESS in 57m 09s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/51e07447242646958fd24e5e87772818 : SUCCESS in 2h 54m 16s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ee1ae209ca024e00a7a84159b0ab130d : FAILURE in 1h 49m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/53efdf09a6e8484f829cab6b3631e7b8 : SUCCESS in 1h 46m 23s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fb7c992003ef495d950564ae77ebdcd4 : FAILURE in 1h 40m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c0da14b8735c4dcc8d037d63a5278219 : FAILURE in 1h 38m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d953ee9c6c98446f8bca0798cecdd5f3 : FAILURE in 1h 02m 06s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/db78d3a551c54b1eb7c1e00bcb4b7a11 : FAILURE in 1h 14m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dfb01a69e1e43eaa73d0d71a36b3e28 : SUCCESS in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c435cabd11d979022b507a351ac00294f676fc4b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 23:13:40.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d570dbc43a0a9c600fef857a6f9b57c7a4aaa58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 01:04:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/701a8739ac1147b88fb1779b98b71741 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9416b49a3bd642e79c4375a53319a9ae : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2972e0ca5f2d4f5ab3a0f394a9d272e5 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6ad013c07e4bba9e5e2500dff02dd8 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3a299122109481f94c2f85604946397 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b2a82ef43a46caa1936cfc40cf5f2b : SUCCESS in 4m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5896973efd3d41248bb7c5cbd969c4c7 : SUCCESS in 8m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9da7da1e0a4411ba9c38568c931dad5 : SUCCESS in 5m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dbd868eacedb45128c7f72c785cd9d05 : SUCCESS in 51m 51s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b03c3fc34a2f4b43b1fac5e0bd46a3bc : FAILURE in 1h 20m 42s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/13d9096af1404e238005386fb6d853e8 : FAILURE in 1h 38m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1d58584e371242a6b8c58dae682dc2dd : SUCCESS in 1h 49m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/04f74255cb74411f95cda6e17c4b73c2 : FAILURE in 1h 40m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/493688ddbc0b428b9d8d9213b2412048 : FAILURE in 1h 39m 42s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d39699ba8ae24ba380c39e1f1a645735 : FAILURE in 56m 15s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/157afc31dd5a4e2689ba217d13a1e606 : FAILURE in 1h 05m 28s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5f00ac1b29f4c20b4d4a61f756c422b : SUCCESS in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e37421cd01633e6c7cc7c6d77b89bf2537caad9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 02:20:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e5970661cae5cc93eeeab7fcfabe42179c60b71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 04:01:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d1d5a9c56447e195bd6df7b909a5bc : FAILURE in 7m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94dbdfb920864c328917e2882c0b8776 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ce26bf3c7e4b749a1727f5ac4f25a3 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac8d215ed684c9bb3655288aca21684 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eb73f22e5b5462282de210eb2c58b4f : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7fce820c8d4e22bd556d72eccc56f4 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/056005d7bbd84bf29a5556f021d23759 : SUCCESS in 8m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e361d9d37d44e5eae7527f74c044a85 : SUCCESS in 5m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9dacd3bc4f1c4a30b127a1f541eded94 : SUCCESS in 53m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7ca6bc1f66f244e58c65b196a135d731 : FAILURE in 1h 10m 59s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5707f1bc5a2c4817b3fbc9d21507e8c0 : FAILURE in 1h 39m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5d489a1494b046e883f0699dd8c32f98 : SUCCESS in 1h 36m 09s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f57ee8565b554eaa804244245d60f8a5 : FAILURE in 1h 38m 08s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/dbc438f8bd15480bbc1a0b542d30ebe6 : FAILURE in 1h 32m 25s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/2d6823e0c1e846a1a5aee66aa257ad65 : FAILURE in 1h 00m 29s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c4aff51044fc49e89b3762b073168447 : FAILURE in 1h 06m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97096286bbad4a1fad985ce32201313f : FAILURE in 1h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4c8af5d7d54af2acdb7a27c7f5b095fe8ff7771","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 05:49:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e865348aa42b376194a7bb118298e5db2fde25ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 08:08:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6384c844ce524f01890c31f92bf3f7ac : FAILURE in 14m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010f9c0ce7774985a4d9d7a885cbfc8d : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf82734fa7e46878ececf68ea7d32b4 : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1def5833199449959f4fcd286ae04ead : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63f837845aa344adba64bc5f53945976 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf15dec20e74487aba4a51bf03a22e00 : SUCCESS in 12m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ecb4aebb67b4fd7bf76647a91361056 : SUCCESS in 10m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2fa037892fd24980a4df918014a88eb4 : SUCCESS in 6m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b58f12ed53c5496bb1c48292f34e4907 : SUCCESS in 1h 05m 46s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d904573a082b4340b58380590ab27796 : SUCCESS in 2h 09m 07s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/05c8f8b66e7d47068c0bce6fb6c434c6 : FAILURE in 1h 47m 36s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fa04ad7bacc148cdb7bac9daa29e5e01 : SUCCESS in 1h 39m 25s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/141055735db4430e8fd0f1fdaf9a7658 : FAILURE in 1h 33m 03s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/353d8a4635764d9084537331e8f6bff5 : FAILURE in 1h 34m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/b0380252937741278940296ac8fa6a18 : FAILURE in 1h 02m 57s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4cea158bb57342e28d1dbe7649229059 : FAILURE in 1h 01m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07d949badfb74d008a97f4cb338c900d : FAILURE in 1h 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84edfabff1e98253cf7a3148b145bca9f62c1924","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 10:48:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"171e987fb21bedfe8857b528e1bf06b95761278d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 13:35:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a5cda9ce92948e48c8809ad2fce8afd : FAILURE in 6m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dbe8938a74948febeefbd026e1c90ad : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061233714976467ea19b370e229cd99c : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/976e5e24f53246b48358add8327cc66b : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49ff07c6e1894175a1aedf6b7db9aadd : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32c7bbea42342788fd63c01c55877d4 : SUCCESS in 5m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3dc4b623cc994a9b9515b8f5776f0c70 : SUCCESS in 8m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d46168712a4a4741ae16c9a4fe0ab718 : SUCCESS in 5m 35s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6772e0e85f014360a6d809886eded25c : SUCCESS in 57m 46s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/613bd4f3c04244e1ad99145098ac676d : SUCCESS in 2h 41m 07s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8b46fa392330489389d3a8cc30f3ac81 : SUCCESS in 1h 40m 35s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6e2d760e776841d8b8116439c5439bb6 : SUCCESS in 1h 34m 00s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4a3d1eb6a8b047eaa4ce9de3166c5f8c : SUCCESS in 1h 50m 36s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/080bdea6931e4a47b63a8ed3bc2dc656 : FAILURE in 1h 31m 31s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/abd058f4282945959d41ad495a022fba : FAILURE in 1h 00m 49s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/131f9e4b4031443aa046bc67bb9f446a : FAILURE in 1h 07m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ed6e16a9735450198eb102fbb0e86eb : SUCCESS in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"acbc98bb65806d226cbd36671b9c408c057070d2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 18:21:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dcff19ac6f01b3b0d868600e2c82eb00a5f651b5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 22:13:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5440512d0e5579703bd558776a1129e92fb0db08","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 22:13:38.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3acc70589fd9e5936a14d6fa71114462ce281469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:51:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ab49790d4d4d30bf3c9c1d5e8e0c7e : FAILURE in 10m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09ddd226af0f482daf2c4dc0d77b2521 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e697f12aa340e28e557bcab81210e1 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36fd5cd71236473fa2b9ccb4e4471a56 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97a673bc3e944906802f7412c762f880 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357dce5b63c74ab7a87c5a08970960ef : SUCCESS in 6m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31f3247a119443a99bd7576faeb8bf04 : SUCCESS in 9m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8dfa7ae589854bcfb747fd454b1a7f5f : SUCCESS in 6m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/95db1587e1fe45d4b9fab10321eb00b9 : FAILURE in 55m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9653e33683924b06abef7801e74e2d28 : SUCCESS in 2h 25m 38s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e712f28f0d8f4c47a46d1a7a8f11c6ac : FAILURE in 1h 15m 11s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2ba7ea6a9832455f91a83820a19e67b6 : FAILURE in 1h 23m 52s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4c329fb7221d40cabafeee5760fe0085 : FAILURE in 1h 22m 24s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1b37fd013e2c4a4c8a54239775c9f875 : FAILURE in 1h 25m 59s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9b6b478c56ce46958586f8e25ce30466 : FAILURE in 1h 22m 44s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/de96a3a33dbe4694b4026e0e840e637a : FAILURE in 1h 16m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9e74797962d486895140fd92aaba2d2 : FAILURE in 1h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"107a6b319904b828f995f67170c80ebfc67524f7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 01:30:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a8caffc98d8a51e507d2229fee2533562abdb850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 03:20:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68510c2d7675458e941c1919e16480fd : FAILURE in 6m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae805841dec14621b739799e4148e398 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bee16ca843349d8a244b77d4515908b : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b7567902a6418d89e70665a5f4ef76 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45fd5a2f6bab4bc8aabd849bff252e1b : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65750cce7f54a329ba772df75fcfd0e : SUCCESS in 7m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/00b469b6f0074ed49db81b31da9e17ff : SUCCESS in 8m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/281144a3cbd14221be9581f1d4635d92 : SUCCESS in 6m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/03d74668028440a093b242c1ff950ca4 : FAILURE in 1h 20m 30s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6c9b30a64b474598b27aa0115acb49a9 : FAILURE in 1h 14m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4f5e330e8d544fcf92ee02a6f8e3fb46 : FAILURE in 1h 24m 43s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1fc7d7a248d1475a9ac7622c6c6549da : FAILURE in 59m 42s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fcdfdb05fc5b449d96f40c00aae681c2 : FAILURE in 1h 21m 21s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7512478a055046a4aa3ff0ff0973fd45 : FAILURE in 1h 47m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7a1bc5a8c6c447f39bfad4f73ea1f603 : FAILURE in 1h 02m 44s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e875dd04ef9b4a6a8d5c75f883d8cf0a : FAILURE in 1h 09m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2563835c858c4c11a409c0ea91d1e1b0 : FAILURE in 1h 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a4783d615874e31a3e02c035bd9f218e8605ee3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 03:40:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"00ae21f302db47e6d3f89709501e6b6b2bc3101f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 06:50:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1451b9d839944e79028c042832d21e4 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69bb038a4fb1485b8956d4eeabe804e0 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e45bcfcd0ac46d7ae7934f73bc0d6aa : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/803b5657249245ad9c0bbaad1b1dd61b : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8627864983b48deb549f2464e25d381 : SUCCESS in 20m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e758b7adb8494cb0856603b1cc08cb : SUCCESS in 12m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/107311a7d28341ce8868c8d73aa6e48c : SUCCESS in 19m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32f04bca4248445084b7aec06bbeee4d : SUCCESS in 12m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9593dab7e59147559d390d716e398e35 : SUCCESS in 1h 29m 42s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/93931d80d2c242268cb50473b304ede7 : TIMED_OUT in 3h 04m 37s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0708f3015f824645be1ea7cb02e83349 : SUCCESS in 2h 40m 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6d45c02c3aa14057a3b26e7d76e3b757 : SUCCESS in 2h 12m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/443e9b084b07482382605ca9dbdd33a5 : SUCCESS in 2h 13m 31s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/24df18a1a6de46fe962c85a70b4fa297 : FAILURE in 2h 12m 04s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/2f7cc95dc63e46d8aaf44807eeb945bd : FAILURE in 1h 41m 02s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/12364e9ce12c469fac0dae69468f3108 : FAILURE in 1h 53m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdf4274a1fb54ca0bce35fe0e8888628 : SUCCESS in 1h 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f26f75464877141b85005c90cdf31b00ec96909","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 08:32:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fda50b908af409f8ec42e4bce1cba230730ac9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:04:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d06e3d6fa644f178e93cf0c89fb9f48 : FAILURE in 11m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d779d92b06545429a9ca6f06c1f7e21 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8057cd22f0d4490881e8ad163fcd6ba5 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9085ed749ab74953921d4ec717b9dd3c : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ece8c4a1c894655a2bbb2de137ca99b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204d820335d44999a70840dbf6c3a2ef : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3efbe69ab3c64f4ebbef1b73740c56c1 : SUCCESS in 9m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b28ea280a7da45799997b0dc5f786c43 : SUCCESS in 6m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d251ac56016b49a9bace5c2675c38f94 : SUCCESS in 1h 15m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/190d498427af43d39a3dd2ea8c79ee00 : SUCCESS in 2h 25m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b8d12b6b2a6049688fd464ddb9b2fe76 : SUCCESS in 1h 52m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c54ead7a7c2549259fc139336cd8198a : SUCCESS in 2h 02m 43s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/60e702c50f1c42d7a970238e0418937f : SUCCESS in 1h 56m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1c216204a17949a9af876f929230c6e9 : FAILURE in 1h 35m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d400c608acb84a21aa1be779a0467305 : FAILURE in 1h 13m 09s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/72f4e2b949de49ba94a59e7fee51b7fe : FAILURE in 1h 24m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/563339f08dbd43d8bbf6f166ee65aafb : SUCCESS in 1h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05862cb211574ccd363feafc12b470e062f05a2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:04:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b99752f8d40a047b8db62d915086906c89f83f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 13:41:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ade5321fcff84ebe81dddfd0bdd36e8f : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faadd091a86e43a79700f2ac90c7b5d0 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3afd2ed4b38c463dbf44ecde09448c9c : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bec6ecbe46ea4c018c085f19b78244d5 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873ea6f1910244acafd50f4fc920218b : SUCCESS in 9m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7c0a4595fd246b9a95e17de9f9dbdfb : SUCCESS in 10m 34s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/761b562770b74347ba548853037fb478 : SUCCESS in 2h 32m 02s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/231f767cced64d47a6d431c05624917c : SUCCESS in 1h 51m 09s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ee579355837741748a0516aadfbd2e30 : SUCCESS in 2h 03m 19s","accounts_in_message":[],"_revision_number":4},{"id":"bb5f7d87f0b8b60d5e372391c175b3d070ca4fe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 13:41:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5661903ba4b00dfe8c880ebef5f170881b9981fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 13:43:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a1830280ba42d882597684413f835d : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"72524f9d856dd715242d9d96116f8238a95c048b","revisions":{"fe3669f52a142c4136c62ed6c102e6b0a73151f4":{"kind":"REWORK","_number":1,"created":"2019-12-15 06:26:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/00/699100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/699100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/699100/1"}}},"commit":{"parents":[{"commit":"a0a10f0b947c63ac06787b490afd0ebecef1477e","subject":"Support HEALTHY status for db instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0a10f0b947c63ac06787b490afd0ebecef1477e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 06:04:20.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 06:24:35.000000000","tz":780},"subject":"Support PostgreSQL 12","message":"Support PostgreSQL 12\n\nChange-Id: Ic1d4982a479c2154d7ddcc807d863e2a37fea506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fe3669f52a142c4136c62ed6c102e6b0a73151f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fe3669f52a142c4136c62ed6c102e6b0a73151f4"}]},"branch":"refs/heads/master"},"d5c93175a295545e327dbcb2af3724373297e068":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-16 08:34:51.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/00/699100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/699100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/699100/2"}}},"commit":{"parents":[{"commit":"e048feb145d99d4584e0a4ed63fbbcd557ddf948","subject":"Remove some unrelated code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e048feb145d99d4584e0a4ed63fbbcd557ddf948"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 06:04:20.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 08:34:51.000000000","tz":0},"subject":"Support PostgreSQL 12","message":"Support PostgreSQL 12\n\nChange-Id: Ic1d4982a479c2154d7ddcc807d863e2a37fea506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5c93175a295545e327dbcb2af3724373297e068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5c93175a295545e327dbcb2af3724373297e068"}]},"branch":"refs/heads/master"},"4c6edf202e1c9306d53c9d862ebc872b83ecf198":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-17 10:48:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/00/699100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/699100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/699100/3"}}},"commit":{"parents":[{"commit":"ec8a754198ac1bbb6954d559ea1d5d679c2de457","subject":"Fix the log related tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8a754198ac1bbb6954d559ea1d5d679c2de457"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 06:04:20.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-17 10:48:30.000000000","tz":780},"subject":"Support PostgreSQL 12","message":"Support PostgreSQL 12\n\nChange-Id: Ic1d4982a479c2154d7ddcc807d863e2a37fea506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c6edf202e1c9306d53c9d862ebc872b83ecf198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c6edf202e1c9306d53c9d862ebc872b83ecf198"}]},"branch":"refs/heads/master"},"72524f9d856dd715242d9d96116f8238a95c048b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-18 22:13:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/00/699100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/699100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/699100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/699100/4"}}},"commit":{"parents":[{"commit":"5472d8a57c7c9e82cca122396fea75b8ef05a64a","subject":"Fix the log related tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5472d8a57c7c9e82cca122396fea75b8ef05a64a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-15 06:04:20.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-18 22:13:32.000000000","tz":0},"subject":"Support PostgreSQL 12","message":"Support PostgreSQL 12\n\nChange-Id: Ic1d4982a479c2154d7ddcc807d863e2a37fea506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72524f9d856dd715242d9d96116f8238a95c048b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72524f9d856dd715242d9d96116f8238a95c048b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
