)]}'
{"id":"openstack%2Ftrove~699868","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I9391b0e50ae37c1fc40268cba566a72fba0ce329","project":"openstack/trove","branch":"stable/train","topic":"fix-ci-stable/train","hashtags":[],"change_id":"I9391b0e50ae37c1fc40268cba566a72fba0ce329","subject":"Fix the log related tests","status":"ABANDONED","created":"2019-12-18 20:45:35.000000000","updated":"2019-12-19 02:46:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"38bb0b37be76b46506949da0632ca7a81fc743d6","_number":699868,"virtual_id_number":699868,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-19 01:04:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 01:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"300449f7a0767f4b142b7f6bca988e93b4eb9f59","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 20:45:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e0afcb4f772d8d01a99d8ae3f171297981aea57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 01:04:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38203ce76f504baeacf0ecd86d15bfbd : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67ed009621a747baa4f91806d0be0880 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0670635d40542d6a2fc8e4055002bb4 : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba79086b95cb4030946c27b366bebc1e : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85601c8670d441d497d6a9d4096d2236 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3efb57117244a509fd3f497805e8a25 : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08e32636cd746b8a2d03237660c14b4 : SUCCESS in 7m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eff78b20b1be454fa7a9cb6fc8c394ce : SUCCESS in 9m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dddc41b676214abfa64d863dd78c51cb : SUCCESS in 2h 56m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a9fca71f27e7487c8bb09cf1d7d21a94 : SUCCESS in 1h 46m 26s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6181990d47cb4a1e82038295c72d6810 : SUCCESS in 1h 54m 32s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/fcad0a55647444c6beda738eab3ba81e : FAILURE in 51m 33s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5c791f7702324b1db434ea3496cc7b7e : FAILURE in 51m 31s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/eef93dc895684bf3b2f546f7ad399eea : FAILURE in 48m 32s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d5faffdb5e9b434cbfcf37a3a2f99af5 : FAILURE in 53m 06s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/82c2228fbbe64596836013e8f103efaa : FAILURE in 46m 30s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/57e1d772a4914b709269e8016eccd16f : FAILURE in 51m 38s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d1ae5f35be4487498306565375dcd78 : FAILURE in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbe63e411b6b989d19fd3dc2025307d722a29397","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 02:46:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d8c8219621ba405f71138fa8c177268a7fe6b70","revisions":{"9d8c8219621ba405f71138fa8c177268a7fe6b70":{"kind":"REWORK","_number":1,"created":"2019-12-18 20:45:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/699868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/699868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/699868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/699868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/699868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/699868/1"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-17 09:54:23.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-18 20:45:35.000000000","tz":0},"subject":"Fix the log related tests","message":"Fix the log related tests\n\nChange-Id: I9391b0e50ae37c1fc40268cba566a72fba0ce329\n(cherry picked from commit 5472d8a57c7c9e82cca122396fea75b8ef05a64a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d8c8219621ba405f71138fa8c177268a7fe6b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d8c8219621ba405f71138fa8c177268a7fe6b70"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
