)]}'
{"id":"openstack%2Ftrove~699870","triplet_id":"openstack%2Ftrove~stable%2Ftrain~Iede80c39bb2a7aaf3c0453bd67f85375fa5ee76c","project":"openstack/trove","branch":"stable/train","topic":"devstack-setting-stable/train","hashtags":[],"change_id":"Iede80c39bb2a7aaf3c0453bd67f85375fa5ee76c","subject":"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP","status":"MERGED","created":"2019-12-18 20:56:56.000000000","updated":"2019-12-20 01:38:24.000000000","submitted":"2019-12-20 01:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699870-1576805753398-7a9322a4","meta_rev_id":"bee7b0123561d194828c4283e60eb3e4eb68bea8","_number":699870,"virtual_id_number":699870,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 01:35:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-19 22:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-19 22:49:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 01:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62906a845517fd3c76c26a63dbd85c95cdc467b9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 20:56:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c812573e7ac5a7fff5ef7a4d899af389203c85e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:55:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0555d19ae1d4077ac8201290bf05b08 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f4ef4a703534c4183eff4ef370fb659 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e318893407bc43ef85a1919c10645748 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1196545d91b4ef89e218ea1164912ba : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d58693cc4d840df96c00b41593e743c : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ad8cf4ce4f54793a88767d75e69f7b9 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619b8ce2230b4f1b9986f2e0a63e79a1 : SUCCESS in 7m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49b310e393dc4813a5b1bfa285c04621 : SUCCESS in 10m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d87857b6260e4f33908b12d367fb6502 : SUCCESS in 7m 17s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/a7455802f8924bdc968ad920874ec645 : SUCCESS in 7m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3de517550af54e89982ecdd73f0beee5 : FAILURE in 2h 32m 08s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6044d17f263d48448369567c2b98e1a9 : FAILURE in 2h 00m 12s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/484f038d43f541f0b6524c4e537844c3 : SUCCESS in 1h 51m 02s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/654b1311b48a4f5e8cabdcc88a895aa7 : FAILURE in 58m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ccdc922655d2468aa6d37098d85cd9a5 : FAILURE in 46m 57s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/755394e4f833415d86c6aebde2a5a935 : FAILURE in 42m 06s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c89a25119c394be5ba80225e9d348f61 : FAILURE in 55m 27s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/ec5c54822fe74ea1a987078c99e3b5cb : FAILURE in 58m 07s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/5d812aceb17f45f1b71b12cd596faa1a : FAILURE in 56m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a58939a084d044379ac3f9a3af038d66 : FAILURE in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87ea4551ccfb94e22a99502cef5b3f2ec844ac37","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 19:02:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3adf8bc9df99f97943196427870a0f7301b67042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 21:25:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37cf8684f9f84a8e95ed8019204f02b4 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/529381b3748e48e3bf1bce05e09ca17c : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfbdca5be6c54f4bb3cbcc512f1976e9 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3505960360784ad6ac24c372ce39a565 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8b10a411064de1a6a90bf5e5828dc3 : SUCCESS in 9m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4b2f2e615a84b11b83dd7bfc8304752 : SUCCESS in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619b250a2614418ea58be7e3bafa81b3 : SUCCESS in 6m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04e293ebe4434377823b4f4c8fad180b : SUCCESS in 10m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36578bd15b094abe92062b6bdae20b27 : SUCCESS in 6m 34s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c8fe3b6467fb47bf811c5bef17d0a389 : SUCCESS in 5m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/84b887249ee346d6af1d26358d9f15d0 : SUCCESS in 2h 19m 12s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7e812bb6e1804f338ff921f22cc94c50 : SUCCESS in 1h 54m 39s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e4922ac008a3464e9f316023f59a1079 : SUCCESS in 1h 50m 24s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/6abe0aa5d3314514ae90d1733bf42ad8 : FAILURE in 44m 40s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6f4bd4b2fbaa45388c2a6eaf6bd2da74 : FAILURE in 39m 30s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3642ec5ebe1246d2993541982e37d56f : FAILURE in 38m 02s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/933e421c7b5649cdae2ca53cc149f3b4 : FAILURE in 47m 22s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/32993cb1dea244879973ce26fbd02a22 : FAILURE in 40m 28s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/af84df2904c848b0aa724c0339b96aa6 : FAILURE in 45m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33dafe1b0638467488dbf23ed7b46957 : FAILURE in 1h 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d4e381e2fc734bb8e2c16e98c01f57624b82496","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 22:49:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6228876199cd587731ab246209532d81f5f32b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 22:49:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d62662ca20ca9631bd58d5bb6ee92f8d1ca26d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:35:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c3fde596b4b43629d13fa1058872beb : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dfbe510ea44e379654536d14b8b177 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12945013d53141ec9a2e86e4dccebf58 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6075ce16838c4ce9992b6dd28a138e3a : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a6b2a31841146009f1d8488ffdfdbf4 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7687b4207384f78b3a7e0880130f6ac : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee6db60b0c8f48b0804c4de5f62a694d : SUCCESS in 10m 38s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/affe45db5a304f7593d9d4ed26e14b44 : SUCCESS in 5m 18s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ec6ff269015648d790cc1751d787c449 : SUCCESS in 2h 40m 19s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/00e678cf11384756800f49a77e9bbf6c : SUCCESS in 1h 49m 41s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3ba3913624d3442c8154982e20ab04e9 : SUCCESS in 1h 38m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c0c3cc8de19b1579745a97f7c2f59131bde54597","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dab77d2fbc13b521852b5d8395770be64e274834","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:38:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8882e184e8c496b849550d82a2f9c4e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fbe6d88e016141b0a5043c26de84b3462103304","revisions":{"23952727ca397ca4f91cbff4e57970025e69f1b9":{"kind":"REWORK","_number":1,"created":"2019-12-18 20:56:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/70/699870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/699870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/699870/1"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-11 01:35:11.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-18 20:56:56.000000000","tz":0},"subject":"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP","message":"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\n\nRename TROVE_DISABLE_IMAGE_SETUP to TROVE_ENABLE_IMAGE_BUILD to avoid\ndouble negative definition.\n\nChange-Id: Iede80c39bb2a7aaf3c0453bd67f85375fa5ee76c\n(cherry picked from commit fc0c4fff4cd56a2dc88d083454c75ac2079f3fc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/23952727ca397ca4f91cbff4e57970025e69f1b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/23952727ca397ca4f91cbff4e57970025e69f1b9"}]},"branch":"refs/heads/stable/train"},"8fbe6d88e016141b0a5043c26de84b3462103304":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-19 19:02:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/70/699870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/699870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/699870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/699870/2"}}},"commit":{"parents":[{"commit":"775ec10c4f259f1075aaa761d1e3f6abc5a4c377","subject":"Use dedicated service credential config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/775ec10c4f259f1075aaa761d1e3f6abc5a4c377"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-11 01:35:11.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 19:02:50.000000000","tz":0},"subject":"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP","message":"Rename devstack variable TROVE_DISABLE_IMAGE_SETUP\n\nRename TROVE_DISABLE_IMAGE_SETUP to TROVE_ENABLE_IMAGE_BUILD to avoid\ndouble negative definition.\n\nChange-Id: Iede80c39bb2a7aaf3c0453bd67f85375fa5ee76c\n(cherry picked from commit fc0c4fff4cd56a2dc88d083454c75ac2079f3fc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8fbe6d88e016141b0a5043c26de84b3462103304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8fbe6d88e016141b0a5043c26de84b3462103304"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
