)]}'
{"id":"openstack%2Ftrove~699874","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I92c15c065edf364b955f2243b3599e0b8cae8b4f","project":"openstack/trove","branch":"stable/train","topic":"train-admin-credential","hashtags":[],"change_id":"I92c15c065edf364b955f2243b3599e0b8cae8b4f","subject":"Use dedicated service credential config","status":"MERGED","created":"2019-12-18 21:32:50.000000000","updated":"2019-12-19 15:39:34.000000000","submitted":"2019-12-19 15:34:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699874-1576769668958-c83103b1","meta_rev_id":"e1176a42c95710e88b095dd62eb2c69affd76bd3","_number":699874,"virtual_id_number":699874,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 15:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-19 08:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-19 08:31:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 15:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cce1c4122f0ce3973949f365c9d3b66cdb1019d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-18 21:32:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"343f67aed17f57be484a5f09f5e13f5d7174191e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:35:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"741aea42be53fee2f6f66bd04474e4abdb0e7f2c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:48:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fc1d27d30236295c3f5fbea496e7d2078e13bba1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:57:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ff7616145de21d3cf96ff6e96e93a892cbaad742","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 02:39:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c4f3665724d655a7f2910e3a2b0bc1a8e017ec61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 05:13:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed328f11a7349c6a8d8070c07d830dc : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/011c4372a39c4649af6d41666f8f692b : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a166c0c20ae84801af6e9b019e7db5df : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3ea5c2484a943c88dd44563e04a869f : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3acb3dbc4361440b87b246da32d267e6 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/853aa68ca46948ea89a23ae5b7364fc6 : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debe8adf02864acf8335ea223c982ad3 : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6234e3f2adb4145ae98850979035aaa : SUCCESS in 8m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bad7cfb4572241d9b36652024480d312 : SUCCESS in 5m 49s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ec1be1ba9a1a484898e1a7d496bd1c74 : SUCCESS in 6m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/58aab0101cd94befbcb8f913fef484b4 : SUCCESS in 2h 26m 33s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4e459932e97047059e936bf473f83a18 : SUCCESS in 1h 44m 26s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fb5ccaecc6cc49aa8ee8fd4bbd328665 : SUCCESS in 1h 49m 53s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/71bac2f785c341b1980dc51cbdf564ae : FAILURE in 42m 43s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/841c34e1d98047b580c18577ad4536ec : FAILURE in 48m 42s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/34c13a3d08b94a9eae4ae8247f311408 : FAILURE in 48m 25s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/cee7003fdbe742c79668ea2e70a84251 : FAILURE in 43m 48s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/d91c19c6c484451e9df0350dcfae17f9 : FAILURE in 46m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/b73d813d088c4b50b16ecba6443064c6 : FAILURE in 41m 47s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc0784499eee474e8d8bd247b657aeba : FAILURE in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"20ba2ca82362e75527aa1fb93c38aa32a1047a4f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 08:31:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"47ee0739cead0dc213d87eefb1ca75dc93406291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 12:33:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e7e1d8118f958a3d1ace5b3bc5041d7e8cecb0b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:34:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"750cda09dd181828bc19970d5c6b7753a0fc2991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:34:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef35e6cfbe5143f09b8c1082e35868fc : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2a06dbaba040f088485659e5aa4d48 : SUCCESS in 7m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/505c7ddfef53480c91fdb9a2c67c07c9 : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f1d8950a3348d18c0445b7b85526dc : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c0cd1208a994cb9af1ac693875ec904 : SUCCESS in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c25788c68f4e87a33f9b2ff9034162 : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d18f14dfff264a9d8ea307994288ac34 : SUCCESS in 10m 17s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/81411110e6a244b9a5b1657693888bf0 : SUCCESS in 5m 41s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b81880358af3454789cc7b332569e7af : SUCCESS in 2h 32m 28s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ac938f737a2143d1b569efd196b314e2 : SUCCESS in 1h 52m 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a3adef34464f439681e00113982a3b96 : SUCCESS in 1h 50m 14s","accounts_in_message":[],"_revision_number":5},{"id":"22551c325192d471b00023465b43302d007573bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:39:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b2741638ca4f9eae3d3c2fd4b4821c : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"775ec10c4f259f1075aaa761d1e3f6abc5a4c377","revisions":{"6778940560383dda9791ef28adf8f3ca08273bd9":{"kind":"REWORK","_number":1,"created":"2019-12-18 21:32:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/74/699874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/699874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/699874/1"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 08:58:46.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-18 21:32:50.000000000","tz":0},"subject":"Use dedicated service credential config","message":"Use dedicated service credential config\n\nThe current admin credential config options are confusing\n\nChange-Id: I92c15c065edf364b955f2243b3599e0b8cae8b4f\n(cherry picked from commit f3bb46542baaed3eac1c0dfd625c1ce17feff043)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6778940560383dda9791ef28adf8f3ca08273bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6778940560383dda9791ef28adf8f3ca08273bd9"}]},"branch":"refs/heads/stable/train"},"c1bbe295fd5f6cf8405ef0ff28fd15442691c9ef":{"kind":"REWORK","_number":2,"created":"2019-12-19 00:35:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/74/699874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/699874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/699874/2"}}},"commit":{"parents":[{"commit":"c8c3b775b535406dc17a4468c834c3a92fd9072f","subject":"Use stable/train for trovestack pip constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8c3b775b535406dc17a4468c834c3a92fd9072f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 08:58:46.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:35:38.000000000","tz":780},"subject":"Use dedicated service credential config","message":"Use dedicated service credential config\n\nThe current admin credential config options are confusing\n\nChange-Id: I92c15c065edf364b955f2243b3599e0b8cae8b4f\n(cherry picked from commit f3bb46542baaed3eac1c0dfd625c1ce17feff043)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1bbe295fd5f6cf8405ef0ff28fd15442691c9ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1bbe295fd5f6cf8405ef0ff28fd15442691c9ef"}]},"branch":"refs/heads/stable/train"},"219df8c19a80dc30901c4b003bea2b51fb82545f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-19 00:48:41.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/74/699874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/699874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/699874/3"}}},"commit":{"parents":[{"commit":"dd69878c9b9531791b36dd7bcad5137e2a16629a","subject":"Use stable/train for trovestack pip constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd69878c9b9531791b36dd7bcad5137e2a16629a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 08:58:46.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:48:26.000000000","tz":780},"subject":"Use dedicated service credential config","message":"Use dedicated service credential config\n\nThe current admin credential config options are confusing\n\nChange-Id: I92c15c065edf364b955f2243b3599e0b8cae8b4f\n(cherry picked from commit f3bb46542baaed3eac1c0dfd625c1ce17feff043)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/219df8c19a80dc30901c4b003bea2b51fb82545f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/219df8c19a80dc30901c4b003bea2b51fb82545f"}]},"branch":"refs/heads/stable/train"},"9ddd66778d4871adf9a233f72686028210b5935b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-19 00:57:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/74/699874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/699874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/699874/4"}}},"commit":{"parents":[{"commit":"aa69c42448c7317194d22e921bab29d52ac7eac0","subject":"Use stable/train for trovestack pip constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa69c42448c7317194d22e921bab29d52ac7eac0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 08:58:46.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:55:04.000000000","tz":780},"subject":"Use dedicated service credential config","message":"Use dedicated service credential config\n\nThe current admin credential config options are confusing\n\nChange-Id: I92c15c065edf364b955f2243b3599e0b8cae8b4f\n(cherry picked from commit f3bb46542baaed3eac1c0dfd625c1ce17feff043)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ddd66778d4871adf9a233f72686028210b5935b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ddd66778d4871adf9a233f72686028210b5935b"}]},"branch":"refs/heads/stable/train"},"775ec10c4f259f1075aaa761d1e3f6abc5a4c377":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-19 02:39:54.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/74/699874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/74/699874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/74/699874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/74/699874/5"}}},"commit":{"parents":[{"commit":"38b6f5ef705c158476375896d0cd38a5a2145e87","subject":"Use stable/train for trovestack pip constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38b6f5ef705c158476375896d0cd38a5a2145e87"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-02 08:58:46.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 02:20:58.000000000","tz":780},"subject":"Use dedicated service credential config","message":"Use dedicated service credential config\n\nThe current admin credential config options are confusing\n\nChange-Id: I92c15c065edf364b955f2243b3599e0b8cae8b4f\n(cherry picked from commit f3bb46542baaed3eac1c0dfd625c1ce17feff043)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/775ec10c4f259f1075aaa761d1e3f6abc5a4c377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/775ec10c4f259f1075aaa761d1e3f6abc5a4c377"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
