)]}'
{"id":"openstack%2Ftrove~699889","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I4c08fe8b90e4b2879eb937aebc9d53789698f58d","project":"openstack/trove","branch":"stable/train","topic":"train-admin-credential","hashtags":[],"change_id":"I4c08fe8b90e4b2879eb937aebc9d53789698f58d","subject":"Use stable/train for trovestack pip constraints","status":"MERGED","created":"2019-12-19 00:24:56.000000000","updated":"2019-12-19 15:36:28.000000000","submitted":"2019-12-19 15:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699889-1576769668141-a23e576b","meta_rev_id":"d95374569fe469a364bbab3d41c3ed03cbd4cf99","_number":699889,"virtual_id_number":699889,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 15:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-19 12:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-19 12:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 15:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b2e40e6bc22d7942b1dbb7afe24ea6f9998aefd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c256466fae85fb4a9396f68c2b66e39d27314372","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:35:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fb87f7f13a846bd2999efc91e19094225e8c503","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:48:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f69e9b09b898c3525b61cfecb451726f8c8f6ab","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 00:57:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d0449a303ce861ebbed0463dd36747c72e989614","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 02:39:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"527d1194b892e9aca2bd97bbff3d407ea3df0ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 05:07:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb7d05e299a4003a41289a8c4714223 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bd4fdba4b704457bd3c4dc28cbe930a : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e993932e7e741ffa7cf5b7a0bc2f721 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dafa3ae853c543d3a2b1be3f4bbed3ff : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7313b67099e149a2a8dcfb82a8cf9e3a : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfc1f6a57efe44a5ba58611e35bfe4bd : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398feeaecd6d408fb1b32f652f96714c : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/959d3d9f6db042f49ba6af896d9f2eb4 : SUCCESS in 7m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b559462bce74323936412cc89a50f77 : SUCCESS in 9m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/661897c1b1a442128079b898740b2bb3 : SUCCESS in 5m 51s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ac9192d7ef3a4cf7b74c26e41cd57aae : SUCCESS in 6m 22s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/bbe7cb564b4743fbbb76afe9a9f19876 : SUCCESS in 2h 25m 41s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/bb165331fddd47a28ba76cfa427b3d48 : FAILURE in 2h 11m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3efdb15d82b14c16866d6164c565ffb0 : SUCCESS in 1h 29m 39s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a32ed206d9d943dd9b9b284d1d05e6e7 : FAILURE in 45m 22s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/813d017abe244871b8eec0a73823d683 : FAILURE in 38m 18s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4a3f44afce8c47338474d400c0d4cd1f : FAILURE in 44m 43s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dacf0fbb18e94aa9aff0a98d171b643c : FAILURE in 39m 34s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/2732a9aaf3f64def9243e83713e82a83 : FAILURE in 42m 39s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/abe940d2036d4aeabcaf6fba7fa7864d : FAILURE in 37m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5448d97d154482ca94c58da79f6e323 : FAILURE in 43m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8bed68f7f8247af9ed2d1df8757df87bcd733adf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 08:31:06.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"81a91775c31e4d807197344580dd1230c573abec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:08:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6285f2c6eae34f55abdeb49c2ac7b2ea : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cf478efac5a4f78b996a31b2dac507a : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3304321a48c04ae987ba55c50385c862 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f266e418c894ebc9cca07f8502e2530 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c12f91dac414d99b802768969c77b4c : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47f3140ddd3042e09e2f95f987e15c30 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8144639648f24a13877d2715cb33a9ee : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502a15a1a3f443c2b68e1a051ea7e020 : SUCCESS in 7m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/127e0d1c9ae44f0d902de78d6807e2ae : SUCCESS in 8m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d3b1a3d404d416d9ca0c31137e78fbe : SUCCESS in 5m 35s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2b1ef4c290f849ee96c07bdd1c94ca9c : SUCCESS in 5m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3eae4019a8004957a9d887a7ca7036bb : SUCCESS in 2h 31m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c3a37175ed534821a15cbc32e6eaab4a : SUCCESS in 1h 43m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/02ca9003965d43028ed14e0cb8fae1c0 : SUCCESS in 1h 36m 41s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7f8ed34d998943ae9fd970e0ca9ca915 : FAILURE in 42m 44s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6fe330b3c7ec448cb07179081af329fe : FAILURE in 47m 37s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4c8c7ed39ce24ddda2dba5a7527f11f9 : FAILURE in 41m 31s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7441af97c41b4d759c409ae4418b277f : FAILURE in 40m 21s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/98e0f1268d39451a9f9272073870ada5 : FAILURE in 44m 35s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/da82392c6535461ea8f106b30d79b1c3 : FAILURE in 39m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e688a6f1a444430bcf8bae8f4c764b8 : FAILURE in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"44344ef33861a6953a4cae21cdc70d39ca6c342a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-19 12:31:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2813b3bd1e2fdaa22e694c436911918b51125d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 12:33:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0413dd9361308dd9c3a1dad7274934928be1b096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:34:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5761f89d82494ea89f4c7d363094a344 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02d7bb424e34e9585683e47b332df13 : SUCCESS in 8m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/523a51eff2794656b25243ea98cfe474 : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eec3145fa0347688ac616fec09dca96 : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85707d860aa646419d7574b56b503cc0 : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96b9d6ea709428a884c5dd993424a44 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a87a7e3e8f43f28bccee04452819e1 : SUCCESS in 8m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71686fb076434aa98cf007d7a1166688 : SUCCESS in 9m 26s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/88ab17ebbf1e4307b5fe1cc87519b3ed : SUCCESS in 6m 05s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f4368764950147da9c70547de90cdf68 : SUCCESS in 2h 52m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3d11608c870d47d8bc9b999768e9503f : SUCCESS in 1h 38m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d73aab3f47354cc5be76274093e16c6d : SUCCESS in 1h 44m 17s","accounts_in_message":[],"_revision_number":5},{"id":"2e29406c258e8cff096f07b9a2c56bb0b677aaaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:34:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f2ba6aae393d27b2ab8c3a0e80fc5d4b52fc6c12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:36:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8672296ff1c84f9485d2c5a668779f1d : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b36988364e7467aa1f0305628a3f9c4 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"38b6f5ef705c158476375896d0cd38a5a2145e87","revisions":{"87f5cfc55d5f4a3d7f5d21e0f958183c22c9e1f7":{"kind":"REWORK","_number":1,"created":"2019-12-19 00:24:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/699889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/699889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/699889/1"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"subject":"Use stable/train for trovestack pip constraints","message":"Use stable/train for trovestack pip constraints\n\nChange-Id: I4c08fe8b90e4b2879eb937aebc9d53789698f58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87f5cfc55d5f4a3d7f5d21e0f958183c22c9e1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87f5cfc55d5f4a3d7f5d21e0f958183c22c9e1f7"}]},"branch":"refs/heads/stable/train"},"c8c3b775b535406dc17a4468c834c3a92fd9072f":{"kind":"REWORK","_number":2,"created":"2019-12-19 00:35:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/699889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/699889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/699889/2"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:34:40.000000000","tz":780},"subject":"Use stable/train for trovestack pip constraints","message":"Use stable/train for trovestack pip constraints\n\nUse stable/train branch for python-troveclient\n\nChange-Id: I4c08fe8b90e4b2879eb937aebc9d53789698f58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8c3b775b535406dc17a4468c834c3a92fd9072f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8c3b775b535406dc17a4468c834c3a92fd9072f"}]},"branch":"refs/heads/stable/train"},"dd69878c9b9531791b36dd7bcad5137e2a16629a":{"kind":"REWORK","_number":3,"created":"2019-12-19 00:48:41.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/699889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/699889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/699889/3"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:48:13.000000000","tz":780},"subject":"Use stable/train for trovestack pip constraints","message":"Use stable/train for trovestack pip constraints\n\nUse stable/train branch for python-troveclient\n\nChange-Id: I4c08fe8b90e4b2879eb937aebc9d53789698f58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd69878c9b9531791b36dd7bcad5137e2a16629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd69878c9b9531791b36dd7bcad5137e2a16629a"}]},"branch":"refs/heads/stable/train"},"aa69c42448c7317194d22e921bab29d52ac7eac0":{"kind":"REWORK","_number":4,"created":"2019-12-19 00:57:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/699889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/699889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/699889/4"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:54:51.000000000","tz":780},"subject":"Use stable/train for trovestack pip constraints","message":"Use stable/train for trovestack pip constraints\n\nUse stable/train branch for python-troveclient\n\nChange-Id: I4c08fe8b90e4b2879eb937aebc9d53789698f58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa69c42448c7317194d22e921bab29d52ac7eac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa69c42448c7317194d22e921bab29d52ac7eac0"}]},"branch":"refs/heads/stable/train"},"38b6f5ef705c158476375896d0cd38a5a2145e87":{"kind":"REWORK","_number":5,"created":"2019-12-19 02:39:54.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/699889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/699889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/699889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/699889/5"}}},"commit":{"parents":[{"commit":"09e3a53737e4b97ce1d617315419d690bbc1e8d8","subject":"Merge \"Support to specify branch when building the image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09e3a53737e4b97ce1d617315419d690bbc1e8d8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 00:24:32.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-19 02:20:35.000000000","tz":780},"subject":"Use stable/train for trovestack pip constraints","message":"Use stable/train for trovestack pip constraints\n\n- Use stable/train branch for python-troveclient\n- Fix the log publish related test cases\n\nChange-Id: I4c08fe8b90e4b2879eb937aebc9d53789698f58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38b6f5ef705c158476375896d0cd38a5a2145e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38b6f5ef705c158476375896d0cd38a5a2145e87"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
