)]}'
{"id":"openstack%2Ftrove~700101","triplet_id":"openstack%2Ftrove~stable%2Ftrain~Ied67c7c243d3f8c644708548cce71ea2aa3cf2e7","project":"openstack/trove","branch":"stable/train","topic":"fix_password_mysql-stable/train","hashtags":[],"change_id":"Ied67c7c243d3f8c644708548cce71ea2aa3cf2e7","subject":"Fix an error when generate root password during DB initialization","status":"MERGED","created":"2019-12-20 03:17:11.000000000","updated":"2019-12-20 08:58:05.000000000","submitted":"2019-12-20 08:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700101-1576832188393-69f755c7","meta_rev_id":"cee701f57ed33b07963ccb2cccadba0ec85c97cc","_number":700101,"virtual_id_number":700101,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 08:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":2,"date":"2019-12-20 06:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"date":"2019-12-20 03:17:11.000000000","_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":1,"date":"2019-12-20 06:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 03:17:11.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"},{"updated":"2019-12-20 08:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34b787f191e7bd5b36c0d4395c2b4520cf8cfcb8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-20 03:17:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e126cd1a967d1eb105f3f03540368f4eccee2e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 06:13:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2479b430778440d29f163fe13cd6c933 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63926043d0c4fa29800ab3615b2932c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bfe77cfb2d64ae794a28eb30f690643 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d0edb7df07741bf808c670358d793a9 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b307800b3454a97bdbe9840f37b7e78 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/469a0311c9e544bda4aee11c68b950f3 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4575f8a340d0490ab9d88a1574641322 : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abdc71074b9a47a89a73c7ccb4467648 : SUCCESS in 8m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f70682cb2f941bfbc5d1a436d64e724 : SUCCESS in 5m 44s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/087838b804b845e4ba36c90135e54351 : SUCCESS in 7m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4e47a0df58ce4f2fa14c02db60555921 : SUCCESS in 2h 53m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e43d896696ac4819998d7e8793adfe63 : SUCCESS in 1h 58m 44s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/eecea63e819747d3b38f8bedcf5c4ae4 : SUCCESS in 1h 39m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/691530235b9f44a39f407251641108ba : FAILURE in 39m 07s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/80f2b10f515842f6b3afc81e4b69de58 : FAILURE in 40m 17s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f49ff8e928fc4a7084c4c00e9e07dbe1 : FAILURE in 48m 57s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b6a0e7c3f12144f8bbe29e950f288fe6 : FAILURE in 38m 59s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0fe3b90012344b3989d2f38f99adb0d2 : FAILURE in 49m 42s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/143082b4fea04d9ba9c210067cc3f5b7 : FAILURE in 39m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8975342a47db4c4aa5c8327e7e67ffa6 : FAILURE in 1h 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83285e02c360cea7c8b40c8ad576057f7d779f03","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-20 06:31:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df0e8aa33be1f8cef6e79869e774cfb2a697e148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 06:31:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b915df4bd53044dde4df367915b6b4fc15b1e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:56:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f04775815f8643c9bf8266a5336a833e : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c95baed85845aea0008c4649e21fc1 : SUCCESS in 12m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ecc0fc862f54876b76972e1ffa5d5cf : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c7f58cce994796a39b01b41c5af40e : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be49503f66d64a39ab40e67b0517376f : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1335e1810a048d5b5d1e6c753c1f0a2 : SUCCESS in 8m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6434d150fae9482ead0a3076d3adf121 : SUCCESS in 12m 59s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/cb827f25153e4d8b91b5141b17c29164 : SUCCESS in 12m 50s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e697a68c2863438ea8ce638f5c2f9256 : SUCCESS in 2h 19m 27s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7897c57356ae4638aa220f73ec205827 : SUCCESS in 1h 48m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9927461c09df443992dd87fd9ccdb2fd : SUCCESS in 1h 50m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f899196493142abcec7b053fd636ca4826d29cb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:56:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c0ebffb0d93f15f8a5db0118168dfbd0f20d4e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77256ceae17943098e59b98c81fb83b9 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83ab7910015c299c486249db1d51fb41c7cd1452","revisions":{"83ab7910015c299c486249db1d51fb41c7cd1452":{"kind":"REWORK","_number":1,"created":"2019-12-20 03:17:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/01/700101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/01/700101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/700101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/700101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/01/700101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/01/700101/1"}}},"commit":{"parents":[{"commit":"560857232af665d49ab030920d9cc7872ccadbd8","subject":"Merge \"Support to build dev datastore image in CI job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/560857232af665d49ab030920d9cc7872ccadbd8"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-14 05:28:28.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-20 03:17:11.000000000","tz":0},"subject":"Fix an error when generate root password during DB initialization","message":"Fix an error when generate root password during DB initialization\n\nThis issue was occurring with percona database (v5.6).\n\nError logs:\n`An error occurred preparing datastore: (pymysql.err.InternalError)\n (1372, \u0027Password hash should be a 41-digit hexadecimal number\u0027`\n\nTested with re-enable root function for both v5.6 and v5.7.\n\nChange-Id: Ied67c7c243d3f8c644708548cce71ea2aa3cf2e7\nStory: #2006546\nTask: #36629\n(cherry picked from commit e84718c7824b4567d8eb53346fc04fee94187e37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83ab7910015c299c486249db1d51fb41c7cd1452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83ab7910015c299c486249db1d51fb41c7cd1452"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
