)]}'
{"id":"openstack%2Ftrove~700466","triplet_id":"openstack%2Ftrove~master~If48ef016c6fb7a875ceabc9f106e4cd85858d147","project":"openstack/trove","branch":"master","topic":"api-doc","hashtags":[],"change_id":"If48ef016c6fb7a875ceabc9f106e4cd85858d147","subject":"Improve API doc","status":"MERGED","created":"2019-12-24 03:53:37.000000000","updated":"2019-12-26 04:08:36.000000000","submitted":"2019-12-26 04:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700466-1577333200704-af565c7f","meta_rev_id":"334662d5c03cb9c5643398b3cdf736037a20bcc1","_number":700466,"virtual_id_number":700466,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-26 04:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-12-24 22:46:34.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-12-24 22:46:34.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-26 04:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4a1759632b68283b96082f05148708aa4dd22a0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 03:53:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2affd0bb2dc36288c2d896544160fafd804bd87e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 04:30:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"668e17c6e6abd18f08e52b6e609bbd604128f23c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 06:13:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"faad98a040e29b452ab4c2c6c7abcb714d673110","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 06:55:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3ec9b45f4036fedd513db8c8199a51bf2b76c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 09:15:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3570f80a6be9434caf9044d015a4e9d0 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6b10787fda84387b88609aa260d64b2 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e0c0829671d468aac58b69c53bad7bf : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b70ecf29c94a88a05fc4d92eb1ad55 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cc7680f45d14f23a586104aaf68db6b : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b02493aeb2bd42838572008adcf4e57d : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f863c43cf145493a8c8dba23b04700e8 : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c5b41b9cc4442a49bf72e3640a747ce : SUCCESS in 8m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59bec660c8bc41e2a736ab6a38e24d3c : SUCCESS in 6m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/637b58c7b6fd4ead8a065767fdba3d2e : SUCCESS in 55m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/96b7aec8a43846a981415e6af2dea283 : SUCCESS in 2h 17m 55s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/149eb9cbac594eec860dde2a0635b772 : SUCCESS in 1h 51m 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/53769f39d244492c8d6342b9cd43097b : FAILURE in 53m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a0c7ed5c81ca44d8b20cc6277d3cbe04 : SUCCESS in 1h 34m 18s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/f02ea133646e436b8ae74b9c3f4eb70f : FAILURE in 1h 10m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9efdc6e1314e4a18aa844229d35a5de0 : FAILURE in 1h 03m 19s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/8076fb1966c74223a788d71a46383fdd : FAILURE in 1h 09m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8c8bf14a2eb4d8eba67dbe5478d7c9d : SUCCESS in 59m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a19dc32e78b491c9695d88f35b6cb50a14240e09","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 10:59:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d350110348e6a4a95fd0401d7d7176dbd5f4e379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 12:51:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/74a90696097a43079c992f1966f58eaa : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a1aa2277a74386adc7abcc871ef67c : FAILURE in 5m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b71730e4994649e484c6d6108ef16349 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02db5a4807a48aa808c6566b44df972 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/675792095fa94b2ca4336bc8b3f3f9df : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d12e056a75d41d3b7fb884559cb053b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6942103e254053877946be672283ec : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d556e1fe29c4275a283127a393a810c : SUCCESS in 9m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4cfe778bb75c4360bfc5addd914ddf25 : SUCCESS in 6m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a6d0b1cb6e42421b9d5fa1ef62cbd9fa : SUCCESS in 53m 38s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d0e8dd78255c4013b69409d17293a0b8 : FAILURE in 1h 24m 34s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f05a48326fe1440995938ba1fcd50a84 : SUCCESS in 1h 45m 36s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1257e4b035334f25b5db8645f369b729 : SUCCESS in 1h 42m 01s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/8ccd50e773324dc39ba33306cc784ad5 : FAILURE in 1h 36m 45s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c9c943fff8a54d748e16027fbbcc2963 : FAILURE in 1h 32m 17s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/479e4fa50c45420fb96b500a3c78ebe9 : FAILURE in 1h 01m 18s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c56c0d7d180846d88757d4c271e117f2 : FAILURE in 1h 06m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9db752575a1445a397c5da16f7c2c961 : SUCCESS in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc86495d30b8d3aa909da6ae94876d90c08e3572","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-24 22:46:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0508a3bf16fa828713562e9ce386c81f02330984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 01:28:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ada39fa8501450a8d79352088529b83 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/237f37645c444d0d8e1db0eb8ede4546 : FAILURE in 6m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8c50fa4c6fe42d188e29e2a4ff09310 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac53dc74b81b40c38d0e912b67ab59f5 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eed97b8d436d486d8b87e78157df9abf : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5539bea858004d5996ef4823e8c1c410 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3979d09ef8844db4bc79c875ba5e6f0f : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b9c13c5c5a894408ab726a584965c84c : SUCCESS in 8m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45b183b3df794f9293450a1c6590bf90 : SUCCESS in 7m 46s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f2cff791d1b34c9fa762822cf275f24b : SUCCESS in 49m 27s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c606d73608354bd08a48cc38b3cdb055 : SUCCESS in 2h 38m 25s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6629f392efc6479ab4942a833311843c : SUCCESS in 1h 52m 44s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0cef699138a44dfba4647db8375b9ff0 : FAILURE in 1h 20m 20s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d3d18d2e17cd4727be4167b523cd9ce8 : SUCCESS in 1h 33m 18s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d6153165e6704829a79b26df3e32eb77 : FAILURE in 1h 24m 55s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/565dc4568d9149fba36025405dbecece : FAILURE in 1h 24m 13s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/436625f73fd94ca1a4c0927191dd7b71 : FAILURE in 1h 04m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9242fe02599f4d6181d608660a19f0e4 : TIMED_OUT in 2h 05m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc12ed482929187bd7cc5cab6d7007ca2800c88a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-25 23:30:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"35ddd3a88ac55c280bf10ff602e31aa16e8ecb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 01:50:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96f78bf15fad4bf6a20637b12bc5f084 : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44a9a8a6cfb74f228bc43fc9c12964f5 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/750d0cc372ae48188130773f9b2c170a : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936ae3e3e5c347399207bfc2fc9f6e60 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a31928d8b64cef8f1c2a0c9c5210ce : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/521a4c424e6a4e7ea5207579e3486c1a : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d61493fab34c639a792f486063b946 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/277326e1ab3a4ff9bb3c7abc8ffd00c7 : SUCCESS in 8m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4849cf7b2aa45168af642a3c25ff6d0 : SUCCESS in 5m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d78ef9e260964f6d8ba8252f86974396 : SUCCESS in 52m 11s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5bb21fa97b65431ead4ad79861c95d00 : SUCCESS in 2h 14m 28s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f9c95977347b4ed0aa0cc0c12990ea6b : SUCCESS in 1h 45m 35s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/cf5c2ad01654429099a5ac2f8d759a58 : SUCCESS in 1h 47m 34s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ce1be2ea34e44455a461bf15eab7e026 : SUCCESS in 1h 33m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/61eee1ba3a954146aa64f16aeedd0ea0 : FAILURE in 1h 27m 44s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5a301a1a5e354e939418d9756c6248b3 : FAILURE in 59m 38s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/355fe508eecb4a959713c4b9b596e67c : FAILURE in 1h 04m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cdb0283e14544f9917a4c647187a1f2 : FAILURE in 1h 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"03e8655550e7b076efd29d177466bc55a26f5807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 01:51:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bacf7e728d8c463cf1c917ebeb8b272e64b03c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 04:06:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6cf172330fe54c678093487c645a3e6d : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ec6bb4207584b6aba18c0de3c9b58ab : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609e8e6467d5448cbceb769ad8a973eb : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3d9999a738545688d049ff2bf1c8804 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf497e2f774448fa910cddae29a95de : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f18a7fcc744bf4a3cdb9ccd1c00ff4 : SUCCESS in 5m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2662c5810e87458ea8e347fcdc669f6b : SUCCESS in 8m 05s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/39512955bd274812aa070f766f608991 : SUCCESS in 2h 14m 44s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7a456c128c884f3bacb044b18f8c204d : SUCCESS in 1h 37m 51s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3e8dd88e760f4cb1b3fc06dddb61caa3 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ad663b3d0dd584b5ffa86f2112804bf63036cb05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 04:06:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2747fd58df5ead380ec71accd0ddaf6711c4e99d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 04:08:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3b123404f68643559fbb180d46e04275 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c782dd092dd4c0e8cdc3ca0bff095db : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dacb400e0bc3c267d55c5d99032b4692f421f141","revisions":{"47fb6434672bba069cb731aaa83068be4dd95c25":{"kind":"REWORK","_number":1,"created":"2019-12-24 03:53:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/66/700466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/700466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/700466/1"}}},"commit":{"parents":[{"commit":"b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d","subject":"Merge \"Fix trovestack tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"subject":"Improve API doc","message":"Improve API doc\n\nChange-Id: If48ef016c6fb7a875ceabc9f106e4cd85858d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47fb6434672bba069cb731aaa83068be4dd95c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47fb6434672bba069cb731aaa83068be4dd95c25"}]},"branch":"refs/heads/master"},"b115ef807d35ab5eb88c80f8d940dde14cdf35a0":{"kind":"REWORK","_number":2,"created":"2019-12-24 04:30:05.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/66/700466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/700466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/700466/2"}}},"commit":{"parents":[{"commit":"b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d","subject":"Merge \"Fix trovestack tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 04:29:54.000000000","tz":780},"subject":"Improve API doc","message":"Improve API doc\n\nChange-Id: If48ef016c6fb7a875ceabc9f106e4cd85858d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b115ef807d35ab5eb88c80f8d940dde14cdf35a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b115ef807d35ab5eb88c80f8d940dde14cdf35a0"}]},"branch":"refs/heads/master"},"1b4d216214ba6e877f2fd51d48b20e05538b92bf":{"kind":"REWORK","_number":3,"created":"2019-12-24 06:13:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/66/700466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/700466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/700466/3"}}},"commit":{"parents":[{"commit":"b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d","subject":"Merge \"Fix trovestack tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 06:13:20.000000000","tz":780},"subject":"Improve API doc","message":"Improve API doc\n\nChange-Id: If48ef016c6fb7a875ceabc9f106e4cd85858d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b4d216214ba6e877f2fd51d48b20e05538b92bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b4d216214ba6e877f2fd51d48b20e05538b92bf"}]},"branch":"refs/heads/master"},"765bb1dc743cebe4b22349d00cfc7f19db3671c1":{"kind":"REWORK","_number":4,"created":"2019-12-24 06:55:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/66/700466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/700466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/700466/4"}}},"commit":{"parents":[{"commit":"b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d","subject":"Merge \"Fix trovestack tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 06:55:22.000000000","tz":780},"subject":"Improve API doc","message":"Improve API doc\n\nChange-Id: If48ef016c6fb7a875ceabc9f106e4cd85858d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/765bb1dc743cebe4b22349d00cfc7f19db3671c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/765bb1dc743cebe4b22349d00cfc7f19db3671c1"}]},"branch":"refs/heads/master"},"dacb400e0bc3c267d55c5d99032b4692f421f141":{"kind":"REWORK","_number":5,"created":"2019-12-24 10:59:57.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/66/700466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/700466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/700466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/700466/5"}}},"commit":{"parents":[{"commit":"b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d","subject":"Merge \"Fix trovestack tox job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4bc1d85f0f6e3a625cbfb0af1ecb9b08dff310d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 03:53:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-24 09:15:09.000000000","tz":780},"subject":"Improve API doc","message":"Improve API doc\n\nChange-Id: If48ef016c6fb7a875ceabc9f106e4cd85858d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dacb400e0bc3c267d55c5d99032b4692f421f141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dacb400e0bc3c267d55c5d99032b4692f421f141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
