)]}'
{"id":"openstack%2Ftrove~702068","triplet_id":"openstack%2Ftrove~master~If15b79661859d5c40f4d9e5fe65e478954b6ddd4","project":"openstack/trove","branch":"master","topic":"delete-datastore","hashtags":[],"change_id":"If15b79661859d5c40f4d9e5fe65e478954b6ddd4","subject":"Delete datastore","status":"MERGED","created":"2020-01-11 09:54:20.000000000","updated":"2020-01-14 02:50:11.000000000","submitted":"2020-01-14 02:47:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702068-1578970078004-d3fc192b","meta_rev_id":"8c28fa1a7474a86c64fb68671e512483b23bf258","_number":702068,"virtual_id_number":702068,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 02:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-01-12 08:30:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-01-12 08:30:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 02:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34e45576415687e2da4f83a8b32da36ae9c217c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-11 09:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e37ebcbe88c200786cb18c6a62231d8429634a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 12:31:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3c133784848044ea99c48a322330abd2 : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a22cc719f424fa193463083c77a4a4c : FAILURE in 6m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e071206146d437997b3bab03bfb3b7b : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959f225b0c75402b960733bfcd8f1329 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5300e637e2c34c678ada10e233b63d03 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb333b874fa74f1da3f3d54bc63d0584 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868e80553a2e49b689bed546806e16eb : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc3706c35c404d2db3e2b5dbef712ee7 : SUCCESS in 6m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7349dcbe799842b1bee63cd24836090e : SUCCESS in 7m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c688a086ce1d40a387ce31542b6ef6e4 : SUCCESS in 5m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0429be57013d4662bc74b3363519c95a : SUCCESS in 1h 01m 55s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6fee34403d8f4267b0234a2aa2124af5 : SUCCESS in 2h 31m 17s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1a4cdf446e7d42b3b5149b355ffdbd61 : SUCCESS in 1h 52m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a28712cea2d842819d96ba6c0f8eb3be : SUCCESS in 1h 34m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a1f4c60f49154d2987750ca68482f8c0 : SUCCESS in 1h 37m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/35eeb3312c374f85b96d54f1e54ebacb : FAILURE in 1h 30m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c96f1b0ca06a4774bb11298bb8618da7 : FAILURE in 59m 37s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/013aff53614a43beb637666cf83e339b : FAILURE in 1h 10m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e83d84f481544a2a6e8a003f79e1d3b : SUCCESS in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49e33379fd4850d7e4927bdac4f3fc8c1011e196","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-12 08:30:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d23a447e0b0c543d2eb698a147453784e3a16054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 08:30:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6a2b6bef990c08c6707db93629034ef29ce977a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 08:40:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a8e9f2822c64cf79b9fcb744a137241 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ca6c86f3d14770a09f45f4069c81c3 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef69c6e11bb43698fda7821851d3204 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/073813b6dafe47d1b62cba340cd94531 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24690826399c4a73bf788cf74ead0f8f : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529a76c99c934c239fa0b9adc41591a2 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90173af4b11c4e519b656399c9d08242 : SUCCESS in 6m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc906cf146fa41589f0025a7d6d51a1e : SUCCESS in 8m 18s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d6106ba5c43142f9878d0d98454274d1 : FAILURE in 9m 10s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/852d437286f84e90a84ce463b9102499 : FAILURE in 7m 17s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/484ab3017d9b4b4ba447987196f5ff54 : FAILURE in 8m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0f9f2e31b66e5133cea42e27ef9b59b94d599c75","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-12 10:41:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ebf70bdab81509f633310faf3951da583d2c99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 11:04:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f5807f1b5a04c01ad6f3871c4d65b03 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb98d4b5ebd4e4aabd1d240a68d7fce : FAILURE in 5m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a732dfc3e18411789e43d91dca16ceb : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e2b37059ee45c891cf519560aef029 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7718d41d4fdf4419904339d8877cf42b : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d68abbccfc524125a8859cdaadab76dc : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26ce903f20747f889a4f5c296020932 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/717a6cf18c22434ea0f752883ad566ad : SUCCESS in 6m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e37b7afc9894b7a82b0cdd1088091b0 : SUCCESS in 8m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf9d049f04c34118886221e9dcff18d2 : SUCCESS in 4m 55s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7f408ed1f8274708825361da2746001e : FAILURE in 10m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c4b6b9550075409d9e55b9cbe54c0048 : FAILURE in 10m 22s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/45d5cf3f788b47be9178fd55e4b3aae4 : FAILURE in 10m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/8df6fed8fb5b41abb69c28b98fd40794 : FAILURE in 9m 32s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/113f7f53ccb94976ba1d9f3d3729a429 : FAILURE in 10m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/6ea6c43999874d6a82c688dd721346b3 : FAILURE in 10m 48s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f66fb337e60140889f9500093c9188fb : FAILURE in 10m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/062bc86dde1e4cda92daa2dec4373b14 : FAILURE in 9m 38s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/467df1cfddba48f980e3df7b69fa5afa : FAILURE in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30cc78ee88a937e59bc507344dceac96bd9536fe","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-12 18:35:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf305f3e877c639ac7198af0a583316385355a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 19:22:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1f083c417dc4fc8b18270d072341c81 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e9a0651a808450c98218a4888f2a7e6 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e046ed5d90f0402ea604fc6b06c332d6 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8545d218761b46d5a66634578a0ebf28 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a45781d5ab7149e9bde63c279ff8eb6e : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1daa8bb843d466f91b0cf61f99633f9 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2982a5cced6474d951b1194376f8bd5 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6f9a6ca4d19464f81bec0ad8cd9f1a7 : SUCCESS in 6m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90131a983a0640e287959d583e2b9778 : SUCCESS in 7m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fc8d25ebc7d4ceb8c390692f009a031 : SUCCESS in 7m 20s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1eef139b20a14bf08b3f98df9a3794c9 : FAILURE in 8m 42s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/aebfd26962a444a4b93e6d13b7f3e90a : FAILURE in 10m 13s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c0665e7941c34dda8301a7fd20c35f25 : FAILURE in 10m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c8fad23a05c149638b3dcb41be82d917 : FAILURE in 14m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/81f4bd2d7e3f4c84ac2e83ad96c08b16 : FAILURE in 9m 18s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/13ade699e1c04998a53a181f250ea8e7 : FAILURE in 10m 25s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/41cb4b91d4ee4811b8502be362fdc79f : FAILURE in 10m 07s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4dc7827e031e424380101ea9d37deb6c : FAILURE in 9m 24s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09aeeb26c0ad4509ad5084b5e4bba3ad : FAILURE in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6c644b08af9be6b26955d267abadf44bbffb9f8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-13 20:56:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77a03670cbedb4cd1904c9376f680f743b6f7b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 23:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f87b6ae7be21426f8d1a82b41cff2272 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2e51611b9594108bc12530a4cd25289 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be4c4bc5382b435e83a9711da1360838 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6218b88d0bf542d8bc454d7371277cfc : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b2af156fb14f43bd42d24b871bedc4 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c80ee2c2260b4bb0b70fe18ac253a6a5 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85120521262544588a0f0002751c5c68 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31700e87be8a4b11abe6893c701222d6 : SUCCESS in 8m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ccc935e2e8d417ca73a39dde5ce32ba : SUCCESS in 9m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45e21fac5b7648e2af83905be1abeb8f : SUCCESS in 6m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0152135ba1d145e4bbd5b197a2054700 : SUCCESS in 1h 06m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/91f762dbdfb0467d9efd9902baa98c8e : SUCCESS in 2h 42m 35s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ae4479640f2e428d8c52bcc20b1b30df : SUCCESS in 2h 00m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/898a01a254a74eef87a261df4334bd99 : SUCCESS in 1h 45m 15s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6b53e1145da44ebaa9a4b47a284ef782 : SUCCESS in 1h 40m 40s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/dc7ee0256de94013bd37e5e4fc47b3c2 : FAILURE in 1h 20m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0df1d0ba3a6d4e53a26f6fc3a317f584 : FAILURE in 1h 11m 07s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/54a703e16e3b41b59c0e382d0c0208c4 : FAILURE in 58m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a4b0f8b761848e89ea73bfa7e6ebdce : SUCCESS in 1h 07m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9a8aab2c10a449e1841830733c2baa1ea85e81f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 23:44:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce546326bd4e0842f330a69b23637fd82f2cac63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:47:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/df19ab2389214ba39c5dff782c180386 : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3361b6ca5134e28b0f212eea4cd374d : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adf8b55968146dd8df98d4c944c6e05 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c94f298d1fa4edea658ef0cfec0c347 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2bde69469de415ea36740ec39fb1f31 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d220179c5123407a8f1c142c826400fe : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a89c635748354ea3ae0cb3b4bd935620 : SUCCESS in 6m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14f93a1b5cf94c859f38e9e3ea6b5d18 : SUCCESS in 8m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/668fca46e0af4bb28fd82bbd419c10f1 : SUCCESS in 2h 41m 59s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7c059bd861744a7681cf489041e8d308 : SUCCESS in 1h 37m 38s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3ed77900616e494b969ae0a350161a6c : SUCCESS in 1h 32m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6d7c230a3174da13a345de8ef6d4b24ca2592402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:47:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d6a1819d290928906ce4300630427e4e194080d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:50:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7047e054175a48ec879a27b6584b5561 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbea297782d9451c9074d3433e096cb1 : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4088a8dd1ea4d4a81ea3dc9e7ae917e : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4551a6cd7c1f27a867460de556fb9185ae42f21e","revisions":{"4551a6cd7c1f27a867460de556fb9185ae42f21e":{"kind":"REWORK","_number":1,"created":"2020-01-11 09:54:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/702068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/702068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/702068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/702068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/702068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/702068/1"}}},"commit":{"parents":[{"commit":"a5f6c9b82ceee093b3e2fcb71ca8db102cabe452","subject":"Fix duplicated words issume like \"object of of the instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5f6c9b82ceee093b3e2fcb71ca8db102cabe452"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-11 09:27:42.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-11 09:54:09.000000000","tz":780},"subject":"Delete datastore","message":"Delete datastore\n\nAdmin user can delete datastore if there are no instances or backups\nassociated.\n\nChange-Id: If15b79661859d5c40f4d9e5fe65e478954b6ddd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4551a6cd7c1f27a867460de556fb9185ae42f21e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4551a6cd7c1f27a867460de556fb9185ae42f21e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
