)]}'
{"id":"openstack%2Ftrove~702164","triplet_id":"openstack%2Ftrove~master~I18654091cc39a0a5de17ff9822f14d6c41facf42","project":"openstack/trove","branch":"master","topic":"trove-guide","hashtags":[],"change_id":"I18654091cc39a0a5de17ff9822f14d6c41facf42","subject":"Add running trove in production guide","status":"MERGED","created":"2020-01-13 05:37:36.000000000","updated":"2020-01-14 23:46:46.000000000","submitted":"2020-01-14 23:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702164-1579045438784-a981c286","meta_rev_id":"7fd55fcb304960d6556738fb0a3e73448585870b","_number":702164,"virtual_id_number":702164,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 23:43:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-14 23:11:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-14 23:11:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 23:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"780ca408128a615b21ef1ecee5c65858daf41cf9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-13 05:37:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69b04f876272b0622b2b7ee6128fa74404da96b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 05:45:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3bd3b5842364376a0741a3d9abd73b5 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e47503de4d4c1d988df471df01006d : FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"76613382f1b88fee4d2a2ad41a453c02e272d86b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-13 08:50:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78190f8b71887ac9b8ebccb5f29c391bd7d1a4ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 09:09:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b7b75f336542efa8a455fb2083c6d0 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9792b80f0aa740e295eba2ad659e9981 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"499dc4bd157d793cf58f9515d36c8272970910cc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-13 20:51:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2801f18f2ebfda7db2f3737a4e110c00009e02fb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-14 22:32:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b71115e8802eb1c3ace2b8b8ec8a12c90f76017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 22:57:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d6811e26744b3facbd3c412639c7e2 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4657ce90fef54d07b89ffff912684182 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ad4e49034c7644ec121c83a5da92783d5741b458","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-14 23:11:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c4a3528fb236814e9ddf2abd234fb287b0c96422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:11:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60c532ead45ced9e7ed4de810525908c5e44d526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:29:58.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5de3a675f6f103ccd563fdebb05e2d314219adc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"87ad2ae55156430af6aa20dce6adae1071741eeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:43:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c957c99f7944f08ba0ace26424fd67c : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9061a6759db743e98afa77855209ed33 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":3},{"id":"c1278e919834ab6c527d9db90ecc37efd54a7c26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:46:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569cea24c85b4fceb62f221b1e20a803 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d57354942eff0faff3c40d1b8dcda7a465ff1d75","revisions":{"4899db4882ebe01e7838a80a5ef06472413364dd":{"kind":"REWORK","_number":1,"created":"2020-01-13 05:37:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/64/702164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/702164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/702164/1"}}},"commit":{"parents":[{"commit":"1191bc8ccd463c8769e67478d64bf5b7847c6eb2","subject":"Config admin clients as default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1191bc8ccd463c8769e67478d64bf5b7847c6eb2"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 05:37:23.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 05:37:23.000000000","tz":780},"subject":"Add running trove in production guide","message":"Add running trove in production guide\n\nChange-Id: I18654091cc39a0a5de17ff9822f14d6c41facf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4899db4882ebe01e7838a80a5ef06472413364dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4899db4882ebe01e7838a80a5ef06472413364dd"}]},"branch":"refs/heads/master"},"c6432f744329abca0d5af1a5a778c67f39aac2ff":{"kind":"REWORK","_number":2,"created":"2020-01-13 08:50:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/64/702164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/702164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/702164/2"}}},"commit":{"parents":[{"commit":"1191bc8ccd463c8769e67478d64bf5b7847c6eb2","subject":"Config admin clients as default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1191bc8ccd463c8769e67478d64bf5b7847c6eb2"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 05:37:23.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 08:50:43.000000000","tz":780},"subject":"Add running trove in production guide","message":"Add running trove in production guide\n\nChange-Id: I18654091cc39a0a5de17ff9822f14d6c41facf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c6432f744329abca0d5af1a5a778c67f39aac2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c6432f744329abca0d5af1a5a778c67f39aac2ff"}]},"branch":"refs/heads/master"},"d57354942eff0faff3c40d1b8dcda7a465ff1d75":{"kind":"REWORK","_number":3,"created":"2020-01-14 22:32:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/64/702164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/702164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/702164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/702164/3"}}},"commit":{"parents":[{"commit":"8356f849f00cbfe92a8b0d380ee2b6432e669942","subject":"Merge \"Check network conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8356f849f00cbfe92a8b0d380ee2b6432e669942"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 05:37:23.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-14 22:31:56.000000000","tz":780},"subject":"Add running trove in production guide","message":"Add running trove in production guide\n\nChange-Id: I18654091cc39a0a5de17ff9822f14d6c41facf42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d57354942eff0faff3c40d1b8dcda7a465ff1d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d57354942eff0faff3c40d1b8dcda7a465ff1d75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
