)]}'
{"id":"openstack%2Ftrove~702325","triplet_id":"openstack%2Ftrove~master~I922a1c5469309704cc6dd60a1ef57e43a98a3c00","project":"openstack/trove","branch":"master","topic":"check-network-conflict","hashtags":[],"change_id":"I922a1c5469309704cc6dd60a1ef57e43a98a3c00","subject":"Check network conflict","status":"MERGED","created":"2020-01-14 02:24:04.000000000","updated":"2020-01-14 08:31:46.000000000","submitted":"2020-01-14 08:25:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702325-1578990355515-17819b5d","meta_rev_id":"996d3b5320142009496169c12d0824fa042c2b3b","_number":702325,"virtual_id_number":702325,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 08:25:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-01-14 05:56:08.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-01-14 05:56:08.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 08:25:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d633e36b6f5022f56510af7752f9f97d7ea04438","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-14 02:24:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61b8849a53ddd460f3c40c7506138416a5e66183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 04:53:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a649232ceed94b72b15fd954238d78f8 : FAILURE in 5m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9374c2ee6951467994dc8210f3459b9c : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f34f76aa734cee938d5f9128e7b380 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/772a0812f94940dd971b5172f5fdbe60 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5a4d5bb0a4c40089777ceba4bf05a85 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a6a4fa091042e8b908439c71c8b8e3 : SUCCESS in 5m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2fa1d9ab281a4dfe9c58fdcf0c4d9b62 : SUCCESS in 8m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49588ce9f44647ffb6668c7266e9dc22 : SUCCESS in 5m 18s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3f813c95fc324fa39e9bc64fdf3b5439 : SUCCESS in 1h 01m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/41a51e8585cb49dbb4fd8e6a9cf4ed13 : SUCCESS in 2h 24m 07s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/266dc57edf2147fa91c667ff18cbd627 : SUCCESS in 1h 41m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/60784052857d4ebf81a5e158314e74af : SUCCESS in 1h 37m 26s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6be6b4cfb10d4bf9bd14cbfae7c9d80e : SUCCESS in 1h 39m 12s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/b9c7f0bcf78947d687b94bc8b80bf03f : FAILURE in 1h 29m 22s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/07548da12a1041e9809aeb2a624aad5c : FAILURE in 59m 30s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c87b2b10c5c44b73a9401d683040b385 : FAILURE in 1h 07m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfe9de5849ac449ea14fe4a7b977561d : FAILURE in 1h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9386312d37f6da340a7ae9cbf681ee739754f0b2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-14 05:56:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d1da515e571087b5b7cdb9e2ac6e94de4684ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 05:56:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69cf794928a97572755101e2123747c51a43eb18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 08:25:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d500a9724a4a9f8bdb54f54c31de60 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f235a8aac6642798c7a0ca187da6ffa : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e4e7d6831b24934a0a040707d77d6a1 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82ff9c66f69849489672fff7f7770b99 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768a5d27b37445b58d34348fde40f6a6 : SUCCESS in 5m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/53f0f5773d284751a04145275944c635 : SUCCESS in 8m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3dd480c2856947e5a7fe56ffabfb57f9 : SUCCESS in 2h 24m 40s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/78d0794bbd924a1e8b5d874318b12152 : SUCCESS in 1h 51m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/df58a8433ac14dc78ad72519d1d26c7a : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb5a5915e55756081ba95ab532d3f81ac3e6a2eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 08:25:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8379ea0949318ce23961f8c3bf24669925128a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 08:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cf07458a574cfc8c247e3a0ee762b7 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e025538fe875b9c388cf13e5f4ba18f677ce2576","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-14 08:31:46.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit cc9eda0242d3e650f17eb9ec1f0ce0a4a76561e1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"263339b4db0fb8e1d10c9933b71a0d5b7140cc68","revisions":{"263339b4db0fb8e1d10c9933b71a0d5b7140cc68":{"kind":"REWORK","_number":1,"created":"2020-01-14 02:24:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/25/702325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/702325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/702325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/702325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/702325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/702325/1"}}},"commit":{"parents":[{"commit":"a5f6c9b82ceee093b3e2fcb71ca8db102cabe452","subject":"Fix duplicated words issume like \"object of of the instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5f6c9b82ceee093b3e2fcb71ca8db102cabe452"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-13 22:15:37.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-14 02:23:26.000000000","tz":780},"subject":"Check network conflict","message":"Check network conflict\n\nThe user\u0027s network to create instance should not conflict with the\nmanagement network.\n\nChange-Id: I922a1c5469309704cc6dd60a1ef57e43a98a3c00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/263339b4db0fb8e1d10c9933b71a0d5b7140cc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/263339b4db0fb8e1d10c9933b71a0d5b7140cc68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
