)]}'
{"id":"openstack%2Ftrove~702842","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I2fa79077660a45787b0e74799264b8533fc4e211","project":"openstack/trove","branch":"stable/train","topic":"fix-bytes-instance-key","hashtags":[],"change_id":"I2fa79077660a45787b0e74799264b8533fc4e211","subject":"Always use string for instance rpc encryption key","status":"ABANDONED","created":"2020-01-16 10:59:20.000000000","updated":"2020-01-16 21:45:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c52d63aa38fb40b63d7f94a7d30f3b23a852886b","_number":702842,"virtual_id_number":702842,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-16 14:12:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 14:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d40ed11b287f4faab7af262b9a69076d89f0eca","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-16 10:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86e60795cef4244eac262285a282525e5bfcecb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 14:12:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/171423c768c94abab69c44da3f9ac9e1 : FAILURE in 8m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb2eeda7e54542a89bdc4ece4b00a405 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e89a447a87469e892a1008d0dded7e : SUCCESS in 8m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06c329dda93e45dbbcafbb1678a7fed6 : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/078e62462d3647f88a86720477f9e444 : SUCCESS in 13m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41b3381087824b90bf243fddd8a08576 : SUCCESS in 18m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfb03df02074a2f902f463e982f3233 : SUCCESS in 7m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b1d6f1fb6d64ac2977ab0f16095708a : FAILURE in 14m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/071800e28ca24662a0eff5d543390f19 : SUCCESS in 5m 22s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d956f58e60fa4261972d6b01281d026b : FAILURE in 12m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/73e5d50835df4a399ef5d7365a728721 : TIMED_OUT in 3h 05m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/82fd716f7ea7486a8478fafe31029131 : FAILURE in 2h 43m 47s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7928b3a1e0434ab29a6aec6e049c3593 : FAILURE in 1h 29m 32s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b24d6e02fb1346179bfc5a06f20bbce0 : FAILURE in 2h 00m 02s\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c198507e94574724b6b0ac1659bad1d7 : FAILURE in 1h 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ccabf9b360ad481a931750a124eb7b536af33c3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-16 21:45:15.000000000","message":"Abandoned\n\nmerged to https://review.opendev.org/#/c/702371/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26500c1085acefdc6e44a64a7fd7fa04557d676d","revisions":{"26500c1085acefdc6e44a64a7fd7fa04557d676d":{"kind":"REWORK","_number":1,"created":"2020-01-16 10:59:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/42/702842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/702842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/702842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/702842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/702842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/702842/1"}}},"commit":{"parents":[{"commit":"baf8fee0a165de39c63d0ae0f72de7222a6f420c","subject":"Merge \"Enable trove-scenario-mariadb-single CI job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/baf8fee0a165de39c63d0ae0f72de7222a6f420c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-16 10:57:18.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-16 10:57:18.000000000","tz":780},"subject":"Always use string for instance rpc encryption key","message":"Always use string for instance rpc encryption key\n\nThis patch fixes the bug in trove-guestagent:\nAttributeError: \u0027TroveContext\u0027 object has no attribute \u0027notification\u0027\n\nMaster branch has already fixed in\nhttps://review.opendev.org/#/c/697225/\n\nChange-Id: I2fa79077660a45787b0e74799264b8533fc4e211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26500c1085acefdc6e44a64a7fd7fa04557d676d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26500c1085acefdc6e44a64a7fd7fa04557d676d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
