)]}'
{"id":"openstack%2Ftrove~709303","triplet_id":"openstack%2Ftrove~master~Id89c664384c3ad443d701f5d4135a9f3f9f744d7","project":"openstack/trove","branch":"master","topic":"cleanup","hashtags":[],"change_id":"Id89c664384c3ad443d701f5d4135a9f3f9f744d7","subject":"Cleanup docs building","status":"MERGED","created":"2020-02-22 16:46:09.000000000","updated":"2020-03-09 10:05:13.000000000","submitted":"2020-03-09 10:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709303-1583748205537-bd11f162","meta_rev_id":"21e2dfb066c1abe573d6c75cf7a4d423bc1daf3e","_number":709303,"virtual_id_number":709303,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-09 05:15:07.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 10:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-03-09 01:42:20.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-09 01:42:20.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 01:42:20.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-03-09 10:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8d9533a985bc2751c34845d37d70a6055e1951f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-22 16:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2718c46598768c6262790360a7a27e8e241d3d0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-22 17:07:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6f7a163045304b9f997f279d465025d534788af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-22 18:11:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"556b71747362611ee37ca5cec1e3673152b25c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 20:24:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/908ac87ec70946f9a63fedf683b64b13 : SUCCESS in 5m 52s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e06f4244fa214fd681e7af91a0b415d4 : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/07eb72143ce64bb3bfeec680fb5fa671 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11bfe3d1bf9e43bdafce40b9d6002d88 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9296a459e55442478cd44928172b0c10 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db1988ed6764981adc2fc48b17bfcb4 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e595e014eaba4226adcc62832f686753 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d51528e415a4f3b8f5fddae938069bb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/573334db4a20420c9dcb3f5a38d39889 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c578d3acc6e546d48782fcd741d79915 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91fbfeaa2b544b1c9d12e3b3225a75e5 : SUCCESS in 7m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/76e3ce0a3d3b4bd899e3e0d7e576d254 : SUCCESS in 9m 03s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8dd9d114e3714633ba22533c915d7465 : FAILURE in 1h 19m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/18b2423aabd2442c8b5a279248cf446f : SUCCESS in 2h 11m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ae715004bfce4104bcb679144ae9e7c4 : SUCCESS in 2h 00m 47s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9bc6a4159a6644c1b02c503116114c6e : SUCCESS in 1h 32m 36s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2c3d659fee57472a93fafef1c9e9f140 : FAILURE in 2h 03m 23s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/24855b2c28bb41a5aea880b1220d5f82 : FAILURE in 42m 26s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4312680d01024531a194e81153bcf64b : FAILURE in 46m 32s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f5f0d7139e5b41898e280110e9ada29a : FAILURE in 1h 03m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51dc5664957d49d182370b7871abdb9c : FAILURE in 1h 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"608019745e86776b263554b1101e32d606b5f1d7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 08:24:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4b3f7ec34d8a742e7b5e204f2cdc5950f7de1546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 10:50:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d502ba28236743aa9f64012042adb984 : SUCCESS in 6m 33s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20c498cb0ce348e28c18b1c494561cbc : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/77c320c2e9a34561a28261cd3cfdb9eb : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc05a9d213d2417ba110356f199ad41e : FAILURE in 6m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4084a9df6631404b9aff4fbb13970ead : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffc30b4b1c443d3a39f5bc229aa9fcd : RETRY_LIMIT in 3s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22add807346e43cb82a7a781ac561f00 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efb2640f7688451ab62e96fe1b201f44 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b55febcb5d2d44988aa2e7732c774cbd : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ba27aa77d4451eba18002e7edb7036 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a68b33c7c3ee49ae9e734513ba4d8bae : SUCCESS in 8m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/adf5a82fec1744c8943091ce7339d81e : SUCCESS in 10m 50s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/44432e6cc2ff4a8e8f875add493133f1 : SUCCESS in 1h 04m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f31b8bf98f1948e8b8c3b46ecc4b5971 : SUCCESS in 2h 05m 45s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fc50b74d67c84cb4b277deb7bb616c34 : RETRY_LIMIT in 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/de3ccb8211624957b8fb70ce1f0c8d54 : SUCCESS in 1h 33m 24s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4dc357c6987243e38b257c58f9af81b6 : SUCCESS in 2h 12m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/6089ffba6bb04fa9ba134a1ec8382542 : FAILURE in 44m 29s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4675101b23ad4c41b82bcfdb9aa52326 : FAILURE in 43m 53s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/810e281dc42e4efb81265edb61bdf60c : FAILURE in 1h 08m 37s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad1aaf23bbc743a986019fa88fc8242f : SUCCESS in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a5f353c5d10f860ae603286fbdb1db86adfdd8c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 12:14:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3505405ba96568ee204f400f95f979b75691d0f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 14:49:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/785de64410b74d039ca3efb0002e50f6 : SUCCESS in 6m 40s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b98dfcb064f4b5589324ae9e23d6ab0 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afd21a138b734ae78aed9909e4ea2c81 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46bce3d9e39a4659a994529ad779c241 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c3ff47de98c4285ae96a5d7d3841bde : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f92a6335f7491bb97723c74eb2c7ec : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8685f732b75e449f95cabb962fb5e133 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efaeba5bfc3147c593ada8008572f4fd : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb634a437a67433b934ddf0d031ef42a : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8437f2090aa140318424b7f43bd7bf6b : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e5f36c624634489968034e1730fdcd7 : SUCCESS in 7m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c41d6adc7ec4a23a6052e971bec15de : SUCCESS in 8m 13s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8ba6091cf11446e989b8fbd513448a32 : SUCCESS in 56m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b8dd2e8dd7e7417186c779c4a06f222e : SUCCESS in 2h 22m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5149283365d04797b4a6278af1e63b8f : SUCCESS in 2h 11m 29s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fe9708295dd3422fa49c8fd85a12b391 : FAILURE in 47m 30s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/257849ae70974eb7893524d43df7f509 : SUCCESS in 2h 06m 11s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c1045b524b4246af9f73e3d2ebfeacda : FAILURE in 46m 15s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/51d9067d3be44361a47fbd8b67cfb8b3 : FAILURE in 45m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d7bd363e51e0431293c9e0aa0a0b269e : FAILURE in 1h 02m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28cff3f747be4098bcbc69289f8e351f : FAILURE in 1h 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14b46990c10c56bb982e5f6db48e5a8714d0e35e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 14:56:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9e246fd2b1b62040d2ad8998a208ee12f6ba0396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 17:29:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45575d6dc8804213a20c2d914bb6baf7 : SUCCESS in 8m 43s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a2f1c2d56c54f089c539e2c3545f02f : SUCCESS in 6m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93d3aa14b2f74825b4bd8ea1492be5d9 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81da88860fc44e4295be04525ed59b62 : FAILURE in 11m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7621c06152034330aa0892243c70aefb : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3968b42e795b4d1e916413cbb695e9cb : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a35a7008b4594be287a2777d049b262a : SUCCESS in 10m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a3fbe09ecca43e69169df1d19ce77fa : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58233368d60f4f6996776df75cd6b40a : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a8475cf23b4834bda6ed2abae45b6a : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2de75e7e174c481db8f748930bbe8454 : SUCCESS in 8m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f00edd92b3af49fba63b5ab160e1ab82 : SUCCESS in 10m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5f6c689963544a94a09c7144c0e4414a : SUCCESS in 1h 00m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b7f2eaa75a8f4d5abb7a550b7dd879ab : FAILURE in 51m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5b9316f9921c4d149d7d74cf24b919f4 : SUCCESS in 2h 29m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9a0ecc401fa047e79fae0d3c47a34a46 : SUCCESS in 1h 25m 44s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/54efd0a575614ec3a2a12f4f508eab82 : FAILURE in 1h 56m 14s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/00df45f386d04ef6a81e54b165b17a8b : FAILURE in 42m 58s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/a5264e99641e4b81a8dddb5882d39602 : FAILURE in 46m 42s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/2b73a4167ccc4bcf9bd0c28bd6434324 : FAILURE in 1h 04m 55s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0cb977d772644daa4e2137fa4c3a009 : FAILURE in 48m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f3da19f0d484c21b888c706b4cb8f7a210d8345","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 17:31:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51c1be416c444d5cef5e940be1afc45fa69034d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 20:00:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a6bc3d2120384f549f33c8f8d44d7b27 : SUCCESS in 5m 59s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f9c4780e10384b4d9821f2518636e3db : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e620be23a1fd430a8341280618be349a : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10655f66d9d4486d8a30df4461be0ff0 : FAILURE in 6m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/518cd462be074a6dbce24c54debbd7ea : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69338fbcea284ef1961f495c3efb58ec : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/026887c223534969a383ad6a3dc0c412 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cefc56ca20ad4e2a8e42e497739cee93 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cc05e64df644cdbbe215302f8bcf95b : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1591fdfd1e874bfda3150369fbc308c9 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d94767459f49b9a5f471ba3b49312c : SUCCESS in 7m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b744f3f1c844138b5ff1a5b5bc6daff : SUCCESS in 9m 25s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4c38c67f291a408998f057d7c7426e03 : SUCCESS in 1h 08m 21s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5312f81413bf44d9bbc92bb030e17355 : SUCCESS in 2h 12m 57s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a23ca57da3c641e38fa2d9a1e2a8169d : SUCCESS in 2h 16m 42s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/678bd1fb8dd54cffb8a7a1310036a350 : SUCCESS in 1h 39m 15s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6157e8b974a34ceba19974cc4241b218 : SUCCESS in 2h 11m 43s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/cc20c061bf8e438d9dfacb95522134ba : FAILURE in 50m 17s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/279227e028bc4d6fb5b24a0076ecaf68 : FAILURE in 44m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/78510df74ab9434e91f6086fe8d19392 : FAILURE in 1h 08m 25s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb87270bc2ed473986b7cd52ea3cc2c1 : SUCCESS in 58m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"448c51dccae648bdf8ddbaf360eb81d1c9ad9dd0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-09 01:42:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"43dea7456f5700232a253f5c062f2bf97ac7ca70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 01:42:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a0703a67ed31065eea2da28ebc03620ef9d1ba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 03:24:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62f90635ad8b47f9aa96b4106c43bf04 : SUCCESS in 6m 22s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e13c4cf87a0467499254aadba903db4 : SUCCESS in 4m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0c97e9ca4c14fbda3ff8ef1c16bd57b : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8de6503d34d048a697a4281a603df171 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873cbe4984df4d14b867ed27f374755a : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a67ca4dcdb44bb68c9c9e32347cc444 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/375486a8d9d24bc4979f804b627cc2af : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723521bb9831488f837e428d870af79f : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee54b6b21f1b4a46a520bc95d8c00d1a : SUCCESS in 7m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54ec345f3dde490b96efb26ed4bef209 : SUCCESS in 7m 59s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0e138763c0cc4309a22f35d3054b5eb7 : FAILURE in 57m 58s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e1ded55e0bc64737861eb572288a80ed : FAILURE in 59m 01s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/83b69f111610424e95a9d18381616a13 : SUCCESS in 1h 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7f62c10ef34f15d2f9136b2f2dc5654d056761af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 05:15:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fdb13960cee5bb761a6c508e425bf1afffec99d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 07:48:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7c5f7eda4e1a4feda436e0db5438ea60 : SUCCESS in 4m 46s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51cb340267824311bbcb35d8fe6dd2af : SUCCESS in 7m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d55acc34adbd4ed3ba45b48bac857f1f : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38181b19a7fc4cb6a5697a444e173f0c : FAILURE in 5m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6468547ca25c465f82628ef3ba8c2af6 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96639df3b964cf1a1479950bf6904a6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d4c0538abe48a9b6ca0f84a4f012e8 : SUCCESS in 11m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf621eb189a4021817dbecb95425f47 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e6563c4dc5c460ca1d779283fc19e30 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9880ea48d7dc45d1890b8a685796f80e : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1cec9359a644d18ae8b9537fd50abb : SUCCESS in 7m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5ed8bcf8c9c42ab98bdef5e8639ed5b : SUCCESS in 10m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e8638b93a78746438475c3289b9bda08 : SUCCESS in 59m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/118fd2616ddc4b32897b62655ce194ad : SUCCESS in 2h 05m 53s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3346008477dd498180edb699e4c78ed7 : SUCCESS in 2h 20m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/86d7a65e97d64187a2f6ca1d61a05dfa : SUCCESS in 1h 31m 02s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f79ef3b6a1144fbb8dcd3bcbf6e694a1 : SUCCESS in 2h 05m 28s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/15e4480a4fa7451788188c8af17482d1 : FAILURE in 46m 14s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5b95b43baef1467298d383c2b561b7d6 : FAILURE in 55m 42s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7849cad3303c4c959ac93200ffa8eb8e : FAILURE in 1h 02m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4692ad22ae504efba4bf9531d26aecc2 : FAILURE in 1h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"151cab7bb20d93dfc7d0fafa8155f155f676e388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 07:48:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a32320f627776d31ebe1b77a369e1c3e25a0275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 10:03:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d143d555eab14ee7a3a015fa265a97a4 : SUCCESS in 8m 31s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27b26f8228434ce796abf3aa0d985ba7 : SUCCESS in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/741a61e0192a47838dbb59b459e31938 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30d00a490fce416f9137d764be6f2def : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e340ab49fba4e0b999e87fb0ff42d96 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9ca54c3918b47d1ac5bcc510151e07f : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4a3b3b4878d4e9ab65cec8ae91480cb : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9bf6dd102e4e10be1da8b5cd2551bf : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e68505ec88452283db175275d56e5e : SUCCESS in 7m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7456a1d880a4af0ac6d1361f601a80c : SUCCESS in 9m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/14c9022ba6934dd7a9b1daa67ea2ccc2 : SUCCESS in 2h 03m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/223304029f3442ccb2739802b4a51bb7 : SUCCESS in 2h 09m 01s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a7868b02236847808338caed0411e86b : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4a052c6f22bacf280feb06cfc6bfe717ece0c880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 10:03:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8ae5c24d86d9d14dbb8ff76fdf22d6ff33f778a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 10:05:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e2ce557943e0440fb1ecdca5fe2f11cd : SUCCESS in 1m 25s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45aef7416a9f443ab0fd647855953171 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c6d2f593e349338068c450ac1b15dc : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc1ccae62961b80358c740662bf482e047f78e67","revisions":{"2f038986fc2f4700b5bc0c6bf473a773df6561fe":{"kind":"REWORK","_number":1,"created":"2020-02-22 16:46:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/709303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/709303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/709303/1"}}},"commit":{"parents":[{"commit":"2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1","subject":"Merge \"Improve the function tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 16:42:31.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 16:45:55.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\nAdd doc/requirements.txt file and move docs requirements to it as modern\nway of building docs.\nRemove build_sphinx from setup.cfg, it\u0027s not used anymore since this\nrepo uses sphinx-build directly.\n\nRequire newer openstackdocstheme version and remove settings in conf.py\nthat are now default.\n\nRemove doc/source/install/conf.py, it is not used anymore.\n\nChange-Id: Id89c664384c3ad443d701f5d4135a9f3f9f744d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f038986fc2f4700b5bc0c6bf473a773df6561fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f038986fc2f4700b5bc0c6bf473a773df6561fe"}]},"branch":"refs/heads/master"},"3ca56b9b58ca042aa38113e4f3ae254eaebc6753":{"kind":"REWORK","_number":2,"created":"2020-02-22 17:07:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/709303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/709303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/709303/2"}}},"commit":{"parents":[{"commit":"2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1","subject":"Merge \"Improve the function tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 16:42:31.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 17:05:55.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\nRemove build_sphinx from setup.cfg, it\u0027s not used anymore since this\nrepo uses sphinx-build directly.\n\nRequire newer openstackdocstheme version and remove settings in conf.py\nthat are now default.\n\nRemove doc/source/install/conf.py, it is not used anymore.\n\nChange-Id: Id89c664384c3ad443d701f5d4135a9f3f9f744d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ca56b9b58ca042aa38113e4f3ae254eaebc6753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ca56b9b58ca042aa38113e4f3ae254eaebc6753"}]},"branch":"refs/heads/master"},"cc1ccae62961b80358c740662bf482e047f78e67":{"kind":"REWORK","_number":3,"created":"2020-02-22 18:11:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/709303/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/709303/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/709303/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/709303/3"}}},"commit":{"parents":[{"commit":"2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1","subject":"Merge \"Improve the function tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 16:42:31.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-22 18:11:19.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\nRemove build_sphinx from setup.cfg, it\u0027s not used anymore since this\nrepo uses sphinx-build directly.\n\nRequire newer openstackdocstheme version and remove settings in conf.py\nthat are now default.\n\nRemove doc/source/install/conf.py, it is not used anymore.\n\nChange-Id: Id89c664384c3ad443d701f5d4135a9f3f9f744d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc1ccae62961b80358c740662bf482e047f78e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc1ccae62961b80358c740662bf482e047f78e67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
