)]}'
{"id":"openstack%2Ftrove~709945","triplet_id":"openstack%2Ftrove~master~Ib52e091417eeb37607c600ac7389cfe41be4e23c","project":"openstack/trove","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"Ib52e091417eeb37607c600ac7389cfe41be4e23c","subject":"[Community goal] Add contributor and PTL guide","status":"MERGED","created":"2020-02-26 03:24:09.000000000","updated":"2020-02-26 07:06:58.000000000","submitted":"2020-02-26 07:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709945-1582700712517-dfe5b5ea","meta_rev_id":"395f521cdac8775b16cfc4d0c50cb1c03ee4704d","_number":709945,"virtual_id_number":709945,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 07:05:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-02-26 06:48:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-02-26 06:48:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 07:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91c5fe9b5e18157e082e02f64b48d64f11f85d74","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-02-26 03:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46922f40906e21f40b6b7555a9e5899620b7ffd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 03:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378e4e4ac5634ed9aec1ee703f27ba98 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242e9542ff7e4eeab954d8872bb01215 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5d0d950ac67c5979cec1db13f616eabcee71c544","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-02-26 06:48:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"738d43075bb17bdd33a0ba91bc2c1f800622f18f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 06:48:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bdd9c5d4d2814ecf87d25ea50c030da78eb235e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 07:05:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed401c64dcce46d39c34030e8755d5eb : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190c0ec96cfc4531904a7e291fc9ea98 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0ae0ef962e555d1ca28a04135f72b53cbd19afe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 07:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd01aaa6e0eb57ddfb634327e17edcdfeef52fcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 07:06:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8588b04620fb477ab632d91583b1ea58 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f1a351f04924f603a7bdaa5cd6d2ea0554bd67f","revisions":{"6f1a351f04924f603a7bdaa5cd6d2ea0554bd67f":{"kind":"REWORK","_number":1,"created":"2020-02-26 03:24:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/45/709945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/45/709945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/709945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/709945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/45/709945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/45/709945/1"}}},"commit":{"parents":[{"commit":"2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1","subject":"Merge \"Improve the function tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d301d0a21863c6c0fbb9e854c7eb8ad8f19bbc1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-02-26 03:22:58.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-02-26 03:22:58.000000000","tz":780},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nChange-Id: Ib52e091417eeb37607c600ac7389cfe41be4e23c\nStory: #2007236\nTask: #38568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f1a351f04924f603a7bdaa5cd6d2ea0554bd67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f1a351f04924f603a7bdaa5cd6d2ea0554bd67f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
