)]}'
{"id":"openstack%2Ftrove~712224","triplet_id":"openstack%2Ftrove~master~I2bd0d81171df85c05423e19ab2e4afa82fcf880e","project":"openstack/trove","branch":"master","topic":"add-troveclient","hashtags":[],"change_id":"I2bd0d81171df85c05423e19ab2e4afa82fcf880e","subject":"Add python-troveclient in requirements.txt","status":"MERGED","created":"2020-03-11 02:22:58.000000000","updated":"2020-03-15 15:53:06.000000000","submitted":"2020-03-15 15:50:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712224-1584287450636-f42c30c1","meta_rev_id":"26e93f30a63f0b81d37a10fddabf3595eb98a086","_number":712224,"virtual_id_number":712224,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-15 15:50:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-15 10:49:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-15 10:49:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-15 15:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e32615490d001cf8f238f08f437d010d31aa2664","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-11 02:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d50f80ba9739c4f57dbb3fe982ec2c7be85a259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 03:33:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/369c3c4b5a8c417881c6786c77763b2e : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcec8d8af49a43fe9b913cf8a5e08f1c : FAILURE in 6m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1a5d55f8d6c4e23b1b8edc4dc6e5ec8 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/034f737622f04da59d5a5a6474f35f81 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d632518ba9444aa6f713e7b9c0b68b : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/898ff650db4f4e32a4469a4bd7603a77 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2038b51cce004bf298077bfdd128d151 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4772f30f6f5b417281362bb2bc8b0ae8 : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/985b914c2c3b4ae38852e9595057e25d : SUCCESS in 8m 49s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1e3fad499db1421faf3aa14dcad69e20 : FAILURE in 44m 39s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/edca061dab0e453691135f0d7056f7d3 : FAILURE in 32m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0d19a694ae91412f865769fe88694012 : FAILURE in 33m 43s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d6e802db3a0d4dda94bbb22970f75f39 : FAILURE in 38m 17s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/0b72b50437b44fd2ba7081e2d9313e64 : FAILURE in 1h 06m 51s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/54e0122336ba4d838c18d010ef230c70 : FAILURE in 37m 08s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0a6887516f5d4792ba91cf02330f591e : FAILURE in 37m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ebee88969b774a728b8195c922e95f83 : FAILURE in 41m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d57218f0b46f4eeb83bd43e6267cdc0f : FAILURE in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a41f8397ed058ab5be279a3cea3dc781f1ee5e5e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-11 22:08:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"633bd9439f42e9c2a6c69eb65903c3fcb1b82614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 23:01:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8253d0f7e7d4bc584b90128559cacce : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/170e0015c11445539076251f8520d073 : FAILURE in 7m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3efabb5fb444869ad24d47c45ae6826 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d271d5ee294935bd2f8c3581dacd42 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55de88f22e224d2c8a096cd90b545737 : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c72bc094e334855819c480555ae7daf : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24e4f01b755c47bf8a49fafb0dcf6edf : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1005ecd62b6415b85ea8d207d45fe40 : SUCCESS in 6m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d03a5c2631c489ca4cdaf07e4b60ea3 : SUCCESS in 10m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/917d7e24e46e42d9905c96eefb49f3e1 : FAILURE in 42m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b084a9d90d5d4755a9d489d440b2dd48 : FAILURE in 37m 45s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b18a624791fa43b99498c33c64f26930 : FAILURE in 45m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1a4256e9d10c4322a5b1abea4404a92a : FAILURE in 31m 36s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4f24ac9d8cf94af0a2bd87928e377856 : FAILURE in 37m 07s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/784a25f05d1e44ceaba14cf321316b96 : FAILURE in 34m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ce736f1d97754d5a94f5102c35597760 : FAILURE in 40m 09s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/194fa311c7cf46c28a6df3f34e704eb8 : FAILURE in 42m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adb0479c7b4d41f5ad56a4cf4f3a517f : FAILURE in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0260b99d2c9e18a2f0989e8e999412e3e8247bd0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-12 03:38:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a85b54c2e54a500a134ab3701364a55e6c6b755b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 04:21:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d48c60db01564bc79715ee2f2b6a92cc : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/023e7d42ed0c418999215a2302d8966f : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb029127147848ddb59bd974db50d7cd : FAILURE in 5m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbbf374e82fa4d449d9fbebc48ae3ad3 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0ccbf8129d47cbacc426a7d1bc2fec : FAILURE in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7833fa4cfc14f5d93d2b461571c62ac : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5add222d1c4936851bed31dc1c7ac3 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1156afb75c40e7a9dcb34e335e0191 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a1aa12ba5849cb83ba66bbd3bd70da : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa1a086aa3142bf938da101fee5b6fb : SUCCESS in 6m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7344ddf3aa374f8e8ad0ce6442da1802 : SUCCESS in 7m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c00a61607c894d258b61a26f525927e8 : SUCCESS in 5m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/35affed372b14ba081b88dfcf025791c : FAILURE in 36m 09s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/32bfce2dee51487ea0039f48537df2b2 : FAILURE in 35m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0672923d239443c99ad1c5fecf81f871 : FAILURE in 33m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ba88614625cb4645beac720b1b62c617 : FAILURE in 36m 53s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/903a6d0e2afc455281c87568214beb5a : FAILURE in 35m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/b5be0873d5634ef2a0b30a385ca714f1 : FAILURE in 36m 42s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/83f95721d1094c6ebf6de9801bfa3538 : FAILURE in 34m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/1ba5720b0de448e9a52502c6ad43032b : FAILURE in 37m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0fe11356c8480e83045f846cb5791f : FAILURE in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38551c38b74bb9d55fe6df9c0b3ee4a03f3b766d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-12 21:00:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa5391756dd116820c0d6758acc1310af984d186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 23:34:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b6a33937ee6b491d9d5007bc0201a42c : SUCCESS in 5m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b197fe7f8e1480fadaab628396663d3 : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ede318c6084098a8be5fe3715f2289 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79090d5e307343fdb0ca4116b1581739 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f98b9271544befad08f3d4695c53c4 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e3c7af503646e0a0e41650473c80b0 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a7f9cbb4d2040659f794e63189880a9 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ded8b6e50cb9470f9197e9a858ff6fa1 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f8150239f6401eb8e0abcb8adf2f19 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8b3da0f4f2f4163bf1f0d98096160a8 : SUCCESS in 6m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5381b3a0d70b41f7a233869ae3e7c44b : SUCCESS in 8m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65967794a04c438a8c7be132a6eccb84 : SUCCESS in 5m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ee5f64148c3142678e6e6bb3570ef131 : SUCCESS in 1h 00m 40s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/db2b4d58a5434764b2eb52e359042ba5 : FAILURE in 2h 09m 51s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fa8069dcbe124180902472b5a14192ac : SUCCESS in 2h 28m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f87841d51a204c5fbea538825bc4826f : SUCCESS in 1h 40m 42s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/99b31dad88dc465daa21a8e87869f310 : SUCCESS in 2h 06m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e8854102e1de460e8f91b64efdb211be : FAILURE in 43m 17s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/69807468a4394bb385e70104780e7d41 : FAILURE in 44m 08s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/609567bd6ffd450da0318220529dc997 : FAILURE in 1h 05m 06s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/495e984e8ce84fa1be2cb7982f195ffb : FAILURE in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56d5d3e58693112193cd6ff64d1e1ff809fccfda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-13 10:57:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25ca0db71203192da09151f2115afcab578cb964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 13:29:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c70a9f247c243f9a316d847565ae3b3 : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/55b7fb9a09884d15bc1b546054cfa8aa : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af9ec0754b746238408ab986c4c9719 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b311ca128e2143588e5b22cb2b1974c2 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f75ac6f446491fa2dc03fbac404964 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9460bb3cae4b40fc8f863347ac5690be : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d571715b6ab4ed2880c364776b35811 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24d64bb3370244cea14812189e983f77 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ab458bc7b14abe9be4e040b1daa6a5 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4e0c0835584017828cf7a85b1f2684 : SUCCESS in 7m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe69472e08044496b95845e13d43d9f0 : SUCCESS in 8m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/947f6b128046439da43738ed0ab79a54 : SUCCESS in 5m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/251eac6d86104a63b97a15ffe93473b3 : FAILURE in 1h 15m 23s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/31e3df5c5bc3438f959ed248bb6e05ab : FAILURE in 2h 07m 45s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/9f5718eacfcc41068b9955c8ed4b3d9a : SUCCESS in 2h 20m 12s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3521e95f48954467affdc4d39a1ae370 : SUCCESS in 1h 43m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/638b8de20e374d39a2cc2b5abec3f003 : SUCCESS in 2h 25m 46s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e83e0b56a4d74539a165ee9f52b43b15 : FAILURE in 47m 44s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/b85a692ad88a46b4a239193956f98e6e : FAILURE in 53m 29s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/9a33e9fde3dc408c9f98d63b5381dc37 : FAILURE in 1h 07m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b918f02fe504965abe0e1c5fa0d27d8 : FAILURE in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d292abb02bacbf45ed3dcf39761b05fc8777623","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-15 04:32:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2e139502e0268a38f2c12e7047a55b8f79eadfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 07:05:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b8e3c160422b40f48d00c3ff1ec503d8 : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8131c450397244c8b568b9ad8d95608c : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9eb5caaf5094b65938bcb278d984241 : FAILURE in 5m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7878d27230443a39d7c4b0409c8788e : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9448f2e4f6f0449d82dc3cc1f15c8e2f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79523cf854f1435abf0a072f793d169f : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b87b0fa30634b2fa1367cde7aa2dfa4 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f72ffa6d7ae04a36a5592e2c58d50c95 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c472eeb05f74e6395e8c7030b4161ec : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8230f9cc05843ae8eb84bc331587a55 : SUCCESS in 7m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/616f4f36b2e24d408550bae218632cc0 : SUCCESS in 8m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b9a76ac73e74a699222df1b453ac85f : SUCCESS in 5m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/453e2e94bce945678acce60f8f4482c0 : SUCCESS in 56m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b74f6aedf37d45d1a5130c1a8e816320 : FAILURE in 2h 05m 33s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c1c18bd77fa4457092c9fa08f2288b1e : SUCCESS in 2h 24m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b38e2b0ec78944bd83727c79081c9f5f : SUCCESS in 1h 39m 49s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c6c3649712624855a1b527a96b817c43 : SUCCESS in 2h 27m 11s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/96fd2baa5b06473589ca4beb35bd0e39 : FAILURE in 45m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/65aa9d6c88ce43779f4847dd614c802f : FAILURE in 42m 02s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/cf91f2ca1b9247629e791076932c7522 : FAILURE in 1h 05m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7e777da07942da8a28a7fb128fa99f : FAILURE in 1h 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3cda1fc733109129cd9f6048ed55b398044e9286","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-15 07:58:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9979aae535917ca97455a17d7b364a18bc30b4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 10:30:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c2e8149d9fb04cd6b810560d342c3de4 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f30b9310b78a4200bbc647b879697528 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/046478c41476463eb8f2882608df1d60 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07c01a92de8e4fb68caec317a1f9a28d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac57013d612248b6a0e8dec681ddabbf : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf2b21e1acf4dea9aef8414fa48f65c : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c35f644c39a419ba104fe3eef6524bc : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0039cb918603498d83cb84d7d63ff1ed : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5737bdf6b6940019b6e1632e5fdae17 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e480d3d69e064420a36ac1c9ea4938cf : SUCCESS in 6m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/083d6a75151b406db2a3adcb7c682fbf : SUCCESS in 8m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebdc07ec8e2f437ab5c98c6ee32ab980 : SUCCESS in 5m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/00b04d384fce4648a089c25b23284ba6 : SUCCESS in 56m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5e16a7dd23514b62b4419fe9b0d55212 : SUCCESS in 2h 04m 38s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cb83c4b22c8e48e7bdaeea1a9d95c001 : SUCCESS in 2h 31m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7cf4f2ec606d4e669b0e8948df4a0dde : SUCCESS in 1h 31m 30s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5dd076c7348d4fedaa565fbf995ed798 : FAILURE in 1h 56m 39s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5dea8bca32c545b9880c361682b9d6ed : FAILURE in 43m 58s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5f9232f13dfa415781abe1621cd8d530 : FAILURE in 43m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/98a32e523487485b98797d3403776500 : FAILURE in 1h 22m 22s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d3e974e10604ca996dfc48b0199c64d : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0491f640e8823ce844b40448c242c7c23a1bfa3f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-15 10:48:57.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3db64e353c8f7a0a78bc3b1ef5f1eddfc2878f3e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-15 10:49:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"41d5c859574352266e216885da44f2278ac7693b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 13:31:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d7f9dbafa04d47bf86b8c6ce91b7cf9b : SUCCESS in 5m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f3d80f8e0bc247edb0f30e355ed6fb77 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a66b2b7efe143c7acef40563ebd6bcb : FAILURE in 7m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74b2e0e12ec2466a8238781ca124377c : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ea6be5fdbd4faaa6f60bdd54f719a9 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1e875315f04beeb45a7af793e096bc : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/286f0d988001422daeef24d0a0e9f993 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfa39adadd994f8ea32f93299d7ce104 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b07e698a9fd49ddae453e2e02c50d22 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5274eee8eb9f40329340872c55633586 : SUCCESS in 6m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd9e2c71d02e46e3a92c2d7f66347978 : SUCCESS in 8m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd1893c309fb4aa7b3ecf8674e34127d : SUCCESS in 5m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/38de65e9234146c9b74beae86f7f565e : SUCCESS in 56m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3df59399a3bb40efb19cbebee7b20952 : SUCCESS in 2h 36m 38s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/72c084c6634e4d1bbedba4d4a6bb03cc : SUCCESS in 2h 19m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b99e7427218049d1b804f290a7662943 : SUCCESS in 1h 38m 17s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/58b81d9d337149f3a81f6aa20b75243d : SUCCESS in 2h 11m 44s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2164992df31d4c748a8c27ba48bd5c62 : FAILURE in 43m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e47e886aaf0541d19fa8d21dd7a0f933 : FAILURE in 45m 12s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/fe4d4039b1d9424e9ce36004f00e50e4 : FAILURE in 1h 04m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a25aaef5685e46728959ab0d87e41539 : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c82241171b8b731ee65beab7572661541c30ab68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 13:31:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03550cabfd57e8b2761b00301054ee2c21d4ba1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 15:50:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"cfc532742b7ae887f30a6f521a594299a94533a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 15:50:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/95001ff68a604afa8247560eb01774f5 : SUCCESS in 6m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7c7222b7f2f4b348c778c6723394aa7 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc908b2b727349cb82e5e8a743d8ff70 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587e7524c1dd4d76a134e9c9d483be65 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d6e2b7f3f74df28de12cb618830e90 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efe3223b44ba49adb80355f2ed4861ea : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18d15c654a3429c8438b98cc95b5cf7 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e13d84817ad4cc4a8e5a608585770b2 : SUCCESS in 6m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4e3d9a45d2e40d0bbe75c99f2fdb152 : SUCCESS in 8m 14s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/35b5fb16255b4bd68ea42fadff3a20fa : SUCCESS in 2h 09m 36s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/72db50c3440e4527a484a70ee709159d : SUCCESS in 2h 14m 21s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c6c1dcd759ce4d2bb19b5351db5f52c9 : SUCCESS in 1h 38m 54s","accounts_in_message":[],"_revision_number":6},{"id":"1151ba3a2b1d8922e6c4e4d046588df681909291","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 15:53:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/276c2ca9a7934bb689467e62fe802888 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c92704090649caa6285aee3659f53c : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a5f43295fb47d3b37d3b61229d1bd5 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d78c74158408ca7ddfc30d6d6b4af7f7ec17a95","revisions":{"c7d8d547d5bbb6628f094a5452aa867180e7a08a":{"kind":"REWORK","_number":1,"created":"2020-03-11 02:22:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/1"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c7d8d547d5bbb6628f094a5452aa867180e7a08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c7d8d547d5bbb6628f094a5452aa867180e7a08a"}]},"branch":"refs/heads/master"},"afb3cf39ee29f580df8f3ef8fbc0500e5047ab45":{"kind":"REWORK","_number":2,"created":"2020-03-12 03:38:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/2"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-12 03:37:56.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afb3cf39ee29f580df8f3ef8fbc0500e5047ab45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afb3cf39ee29f580df8f3ef8fbc0500e5047ab45"}]},"branch":"refs/heads/master"},"674c297eda704fd1c9a28588227d1b84088625b9":{"kind":"REWORK","_number":3,"created":"2020-03-12 21:00:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/3"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-12 21:00:06.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/674c297eda704fd1c9a28588227d1b84088625b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/674c297eda704fd1c9a28588227d1b84088625b9"}]},"branch":"refs/heads/master"},"a2ebcbfbffe4fc0736f9c7bb6153dfb0834a9563":{"kind":"REWORK","_number":4,"created":"2020-03-13 10:57:28.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/4"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-13 10:57:16.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2ebcbfbffe4fc0736f9c7bb6153dfb0834a9563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2ebcbfbffe4fc0736f9c7bb6153dfb0834a9563"}]},"branch":"refs/heads/master"},"7b11270b9511b69dd1ee85bcfa1793d31153b00c":{"kind":"REWORK","_number":5,"created":"2020-03-15 04:32:25.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/5"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-15 04:32:12.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b11270b9511b69dd1ee85bcfa1793d31153b00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b11270b9511b69dd1ee85bcfa1793d31153b00c"}]},"branch":"refs/heads/master"},"7d78c74158408ca7ddfc30d6d6b4af7f7ec17a95":{"kind":"REWORK","_number":6,"created":"2020-03-15 07:58:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/712224/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/712224/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/712224/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/712224/6"}}},"commit":{"parents":[{"commit":"ae2709dcd6ffb7966dde056c6f0bafb20881c594","subject":"Small cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae2709dcd6ffb7966dde056c6f0bafb20881c594"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-11 02:22:47.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-15 07:57:39.000000000","tz":780},"subject":"Add python-troveclient in requirements.txt","message":"Add python-troveclient in requirements.txt\n\nChange-Id: I2bd0d81171df85c05423e19ab2e4afa82fcf880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d78c74158408ca7ddfc30d6d6b4af7f7ec17a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d78c74158408ca7ddfc30d6d6b4af7f7ec17a95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
