)]}'
{"id":"openstack%2Ftrove~713767","triplet_id":"openstack%2Ftrove~master~Ia010941fd9fde2036672f6e6ae8ac488d63c1d4a","project":"openstack/trove","branch":"master","topic":"hiwkby-apache-syntax-error","hashtags":[],"change_id":"Ia010941fd9fde2036672f6e6ae8ac488d63c1d4a","subject":"Fixes the following syntax error of etc/apache2/trove apache conf","status":"MERGED","created":"2020-03-19 00:54:48.000000000","updated":"2020-03-21 14:01:34.000000000","submitted":"2020-03-21 14:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713767-1584799204803-d1a6a7d1","meta_rev_id":"03e4d0cdaf0cfe27e8de4d449a0dea9d98aae29c","_number":713767,"virtual_id_number":713767,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 14:00:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-03-19 08:57:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-03-19 08:57:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 08:57:11.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-03-21 14:00:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"770028a36fcc5cfc1064e345e45c903cdd84e3ae","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2020-03-19 00:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58269efb518bf7d71f65b4d8fa228c0935315602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 03:39:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8422facc30244381942614179d048e04 : FAILURE in 8m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f68d0d7f06b4831bf37e0e107eca610 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8066177bc7a2481ba585d87716ca7e1c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65377b69ea8d4a23ae7fcb1d8dea3976 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e09228595db45bf87124096754cd7e2 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be0159afa2054fe6988ca85a9edd0d7d : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971aa838d41f4760a14f2883df285823 : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e58ef5b37c6349959d178017e866dd35 : SUCCESS in 8m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fb50d354fb1a495bb7a2e7a4d1091cb8 : SUCCESS in 1h 59m 34s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/aa2aae9c43c34369bfdf5377d94f6fec : SUCCESS in 2h 38m 35s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/725af61b821a4703b77389f47ec2b63c : SUCCESS in 1h 37m 46s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/808ea1913c0241599d5a44c5da7511f9 : SUCCESS in 2h 11m 09s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/82cc67ef5b7d4171a01e8ffb3df0b65f : FAILURE in 43m 55s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0afa4440fc6948418b481659e9e40b76 : FAILURE in 44m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/59516ac28b3a47088a360b8bb4dca2dd : FAILURE in 1h 07m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fca156a79f7429ea5e3b9f41b7a86a1 : FAILURE in 1h 15m 20s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/e124eb28862a4ee1a49390167307d3c4 : SUCCESS in 2h 07m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"339c35da15614d3174273048d3241b56323318be","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-19 08:57:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks for the fix!","accounts_in_message":[],"_revision_number":1},{"id":"d7effd03611bc47b26167031e780a0d3da81f5b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 08:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e02d44478d698386faa59c03006724cc0afa7fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:39:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5073d841b3d84cc296d5cedee1b66566 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54199ae72111467ba0079f1fc6e0c755 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/341fd1559c66417b90f9834a57b14516 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8632e599af004377b036302108657391 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d5afdb0a5e4c72b77676e52d918bd8 : SUCCESS in 6m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25998c0a7a7443319b1775aaa59320a2 : SUCCESS in 8m 50s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d42cc822fca0487cb34452f99008d9d5 : FAILURE in 1h 01m 22s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a03020a67ffe4d47b1de38542ec540ba : SUCCESS in 2h 28m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fc315295219c455ab3144f65f5a7a363 : SUCCESS in 1h 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e6ca551b724d54f8b6191229171d082f457bdcef","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-19 20:50:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13188ebf076f30f6e9e7e58ab5bac0d454f9ca6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 23:00:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51a5d2e1d7a543a09258e01aa950ec22 : FAILURE in 8m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12f0ce3e2280463db8e6c0de52dc2a2a : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fe0a73e53942f490d98bf5702bacbb : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea22b78316354b139dc64e9689fe245b : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecf457cfb19540418dca74deb7ea2338 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a92e6bd94662488685fa6bb866ca67bf : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f19f6195ee84bbb9f482988f0eb177d : SUCCESS in 6m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e26f0d2da1243e591a8cdde19f7405e : SUCCESS in 9m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/af4e7b557e4b428595ef0e164b4f856b : SUCCESS in 2h 03m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e25159cbe6f94414ab751565966f33d0 : FAILURE in 54m 40s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/db814a2aef984166a538cdc9f3ba5bfc : FAILURE in 56m 17s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a05bb612ffb94c0f847a7918e28bd180 : FAILURE in 59m 48s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e485ece0478a444fa784365b6376e4e4 : FAILURE in 54m 28s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6b5f75eb38024a3eaa6a928339f6314c : FAILURE in 53m 13s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/1533ff55cdc240dcbae189a18b6c9c55 : FAILURE in 1h 01m 56s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f1cc12b7fba4475b575c103c453f2d6 : FAILURE in 1h 39m 55s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/ef5a07279f5548be86e96725d7ad2d96 : SUCCESS in 2h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efecfb0db4faa4738939873682d9689877e3b201","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-19 23:18:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bce9063dea152035b115cfd5f966ab3d485321b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 01:41:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/145c8e77fdcb48668e034a7d060d805a : FAILURE in 8m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31ee54ed2b8341d68012bfaa7fd017ab : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ababbef11af745ef8a0f4a55f4e83fe1 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7570e1d6698347b9ab805b746e24f054 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01258cd8a30c4b39a775b48164a88a95 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cccc1e08d60b43d39d149d4f03bb894b : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cd4261edfd45e5acdb76cd86865843 : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7decd980c28c49c68549022354185f26 : SUCCESS in 9m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1555c76cf893413a82531751f535867a : FAILURE in 57m 38s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/169ea2a4740f46ae823391b9038b4b22 : SUCCESS in 2h 18m 07s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/faa04a9ffb2c4d38a7d6887eebdc735e : SUCCESS in 1h 30m 14s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/1c80dbd8416c416f8e9aa0b5367c5ebd : SUCCESS in 2h 04m 24s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/27a88fa8987f4b6eaff27d9cdfdf5ba6 : FAILURE in 46m 06s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/2304359e3a5643e093e4b90d5138a1ca : FAILURE in 45m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/8d1a26f21a884052a704844f0b67a21e : FAILURE in 1h 05m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b77c6154f4a4d7d848ce18b551b95c9 : FAILURE in 1h 23m 42s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/5e2c701e6cfd4853a6af9115d8b276c5 : SUCCESS in 2h 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc3c33e35e0245fa308f32e053f59eef305689a8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-20 21:16:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e5824d8cf68c03f8d46933016d8f3ad70a42a92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:39:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a07f4c6d69d404abd82e7e5140f0de4 : FAILURE in 9m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eed3450e618449d9a5fd7201103b300 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79d0527bd314a4a899c59a07c7c115c : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c34f9bcd5a943419273a6dd13f95649 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/766e8412875349d69dc0dd2f2e7ca45f : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c5f4aa73ddf4057bf68be175cc2f81d : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d1bf3b255747038021e47c235cd6db : SUCCESS in 5m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7a8980889664761aea251b5b4555bd5 : SUCCESS in 8m 45s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d5da94748aba4a1c95297723b276a457 : SUCCESS in 2h 10m 37s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0d050294016a45d2abdf2e10a4c6fcfe : FAILURE in 2h 16m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/35d817bfc4174cbaaf5689b84c892340 : FAILURE in 50m 43s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5ae2dcda7f07468b8261eaed21b7afc9 : SUCCESS in 2h 20m 05s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/142fead6458746a08cd58cd0872b6069 : FAILURE in 47m 15s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0b2a6659559844d4a345be9729726038 : FAILURE in 39m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4110fe85041e4fa5b8a49a3df8367d32 : FAILURE in 1h 00m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b9cc3881f2446ebb0470965a0e5815f : SUCCESS in 58m 42s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/bd872216632c4f15a943e2394f560357 : SUCCESS in 2h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce955d744cdc64fe7ca368fbbec5f29a3b94af3e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-21 09:20:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c5cb4bd9b8720d3ac1a030cfcc575e2743f5731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 11:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f982fdceaa0d4341b7da7d9b982c0a19 : FAILURE in 8m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37478e1e3ca643008996256db883fc36 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d450b7b3829648789938f16570bdfe96 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da4b7e6b177341b3955aebe4de32b837 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2e451620c7345339e8519523485c2ea : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/721792b107944448a3021c0885973f05 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af839efe5ac458991336b3b7da53c1e : SUCCESS in 5m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60ac59ce36514be8a9037daac70b606f : SUCCESS in 8m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/851c7f750d11402884ea1c028f5550f3 : SUCCESS in 2h 08m 13s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8da442ada2434103abf27b886b5fba36 : SUCCESS in 2h 10m 52s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0a5c9db79f154bfbb184dabc0fdb4ccd : SUCCESS in 1h 34m 22s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/9ddbae3e778f47469d11fecafd9fcd21 : SUCCESS in 2h 16m 48s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5f4aaa4c3aff40db977c88e4352c5158 : FAILURE in 40m 31s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/48c18169924e449b8d9702385c6a64e4 : FAILURE in 42m 57s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3048357f261b4b3ba52d37644b16866f : FAILURE in 1h 10m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f23c1405b7574b5c9b92a773f0cdd7f1 : SUCCESS in 59m 39s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/70c4e07c456e494e96751ff320111322 : SUCCESS in 2h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f6c4d22af7ae2acc7f904d8eb104871e058e359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 11:42:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2df6e9802c5a1676f16a5c52a96394e34f6eb371","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 14:00:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d76a67143c8f151cb86842dcea68409a28d7d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 14:00:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/772cadfeed9547089ce95ab46d3fe385 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef96fcaa9de4da0aa5806947ec47491 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e3779408a44c00a6dc8b2047910486 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fbc997f68a34e69b0c6654f4f5f0a19 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ed5253338845cd9f7a168caf305a1c : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/682063558df44d049813eb4cd95825dc : SUCCESS in 9m 32s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/77efd910d6544500aaea41291a3327de : SUCCESS in 1h 58m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/12b32276e4f1463e947e6344b918120e : SUCCESS in 2h 12m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/8c9c7fa5d46c44c0baf63b2e6ee3266a : SUCCESS in 1h 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d0ec3bacc11b9ab4679e84b2fe4140b5b6b4b452","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 14:01:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dccf88f081401a9f9bcc7bb85ff78a : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f7f2c743bd470f26d551dce187c8bf0682e0f48","revisions":{"2f7f2c743bd470f26d551dce187c8bf0682e0f48":{"kind":"REWORK","_number":1,"created":"2020-03-19 00:54:48.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/67/713767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/713767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/713767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/713767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/713767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/713767/1"}}},"commit":{"parents":[{"commit":"513e0dd3cd27790718b31c6d0d4b556bd6de0793","subject":"Delete error volumes when deleting instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/513e0dd3cd27790718b31c6d0d4b556bd6de0793"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2020-03-19 00:36:17.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2020-03-19 00:50:15.000000000","tz":0},"subject":"Fixes the following syntax error of etc/apache2/trove apache conf","message":"Fixes the following syntax error of etc/apache2/trove apache conf\n\n$ apache2ctl configtest\nAH00526: Syntax error on line 16 of /etc/apache2/sites-enabled/trove-api.conf:\nInvalid command \u0027through\u0027, perhaps misspelled or defined by a module not included in the server configuration\nAction \u0027configtest\u0027 failed.\nThe Apache error log may have more information.\n\nI refere the manual to configure my trove env.\nhttps://docs.openstack.org/trove/latest/install/apache-mod-wsgi.html\n\nHere are three steps to reproduce the error.\n\nStep1. Install apache2 and mod_wsgi\n-----------------------------------\n$ sudo apt-get install apache2 libapache2-mod-wsgi-py3\n\nStep2. Git clone and Configure apache\n-------------------------------------\n$ git clone https://opendev.org/openstack/trove.git; cd trove\n$ sudo cp etc/apache2/trove /etc/apache2/sites-available/trove-api.conf\n$ sudo a2ensite trove\nERROR: Site trove does not exist!\n\nI files this in https://bugs.launchpad.net/trove/+bug/1867811\n\n$ sudo a2ensite trove-api\nEnabling site trove-api.\nTo activate the new configuration, you need to run:\n  systemctl reload apache2\n\nStep3. Syntax error happens\n----------------------------\nBefore starting apache2, Add \"stack\" user and mkdir \"/var/log/httpd\" directory.\n$ sudo useradd -s /bin/bash -d /opt/stack -m stack\n$ sudo mkdir -p /var/log/httpd\n\n$ apache2ctl configtest\nAH00526: Syntax error on line 16 of /etc/apache2/sites-enabled/trove-api.conf:\nInvalid command \u0027through\u0027, perhaps misspelled or defined by a module not included in the server configuration\npAction \u0027configtest\u0027 failed.\nThe Apache error log may have more information.\n\n$ apache2ctl configtest\nAH00526: Syntax error on line 16 of /etc/apache2/sites-enabled/trove-api.conf:\nInvalid command \u0027through\u0027, perhaps misspelled or defined by a module not included in the server configuration\nAction \u0027configtest\u0027 failed.\nThe Apache error log may have more information.\n\n$ sudo vi /etc/apache2/sites-enabled/trove-api.conf\n$ diff  /etc/apache2/sites-enabled/trove-api.conf /etc/apache2/sites-enabled/trove-api.conf.orig\n16c16\n\u003c # through mod_wsgi\n---\n\u003e through mod_wsgi\n\n$ apache2ctl configtest\nAH00558: apache2: Could not reliably determine the server\u0027s fully qualified domain name, using 127.0.1.1. Set the \u0027ServerName\u0027 directive globally to suppress this message\nSyntax OK\n\nI have tested on Ubuntu 18.04.4 LTS.\n\n$ uname -a\nLinux ubuntu-bionic 4.15.0-76-generic #86-Ubuntu SMP Fri Jan 17 17:24:28 UTC 2020 x86_64 x86_64 x86_64 GNU/Linux\n$ cat /etc/os-release\nNAME\u003d\"Ubuntu\"\nVERSION\u003d\"18.04.4 LTS (Bionic Beaver)\"\nID\u003dubuntu\nID_LIKE\u003ddebian\nPRETTY_NAME\u003d\"Ubuntu 18.04.4 LTS\"\nVERSION_ID\u003d\"18.04\"\nHOME_URL\u003d\"https://www.ubuntu.com/\"\nSUPPORT_URL\u003d\"https://help.ubuntu.com/\"\nBUG_REPORT_URL\u003d\"https://bugs.launchpad.net/ubuntu/\"\nPRIVACY_POLICY_URL\u003d\"https://www.ubuntu.com/legal/terms-and-policies/privacy-policy\"\nVERSION_CODENAME\u003dbionic\nUBUNTU_CODENAME\u003dbionic\n\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\nChange-Id: Ia010941fd9fde2036672f6e6ae8ac488d63c1d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f7f2c743bd470f26d551dce187c8bf0682e0f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f7f2c743bd470f26d551dce187c8bf0682e0f48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
