)]}'
{"id":"openstack%2Ftrove~716462","triplet_id":"openstack%2Ftrove~master~I9014f4ae3083173b49d0ee581c976d06e852a29f","project":"openstack/trove","branch":"master","topic":"trove-vars","hashtags":[],"change_id":"I9014f4ae3083173b49d0ee581c976d06e852a29f","subject":"Add innodb configuration parameters log_file_size and flush_method","status":"MERGED","created":"2020-04-01 01:08:24.000000000","updated":"2020-04-01 23:19:23.000000000","submitted":"2020-04-01 23:17:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716462-1585783067440-28344ecc","meta_rev_id":"7732762a1a92ffb5baa773d045d5c0231d94df4c","_number":716462,"virtual_id_number":716462,"owner":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 23:17:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-04-01 20:18:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-04-01 20:18:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 20:18:02.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-04-01 23:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f678923dc79a83fbde0c27df85eebe8f7d1ae671","author":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"date":"2020-04-01 01:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"128a4f8ce1d0d88570bda1842d27143ee3505d9c","author":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"date":"2020-04-01 01:18:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb2b6be3f4f13baf294f91195c126e765c036547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 03:49:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00ec594b69634f429e40b5d9c970091e : FAILURE in 6m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f313e8200394b5cbe93d1892cb94227 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c705ed286207494cba7c46374bfcfa02 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3957473409fb432ca3b51852b496b252 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5371872c74c2474d90097833a31e21bc : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b914a5eef21463288792194ea2fa301 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac084025e5149ff9648d9910aeb4d8b : SUCCESS in 5m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e88976088d049ea8f9cf2c319816a4c : SUCCESS in 8m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1e86d3d4ff4448287f80aefa7edbdc0 : SUCCESS in 5m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5e27cbc6734540409ac2c4ea2e530340 : SUCCESS in 1h 20m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/05a43c1740b44667a0e3202cfd8ec3af : FAILURE in 2h 21m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/02073d794b9248ad9e716e2d8b38402b : SUCCESS in 2h 26m 44s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5084370df6654ca9b683ab2966987a1f : SUCCESS in 1h 39m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7da2bd1a45384c5b99fbf43ce9f35273 : SUCCESS in 2h 12m 00s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d69c4a1ab80a47b391c3170382044c3f : FAILURE in 42m 49s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0c1c525a11da437e9834e22c01fc9d19 : FAILURE in 43m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/463cdeb17a3c41f18eb7ffa15f2f856e : FAILURE in 1h 56m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19898a2d12f94a21aa793b04fd0eb419 : FAILURE in 1h 10m 13s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/5763c0a6ae0d4847919c02e9e3c8b3d3 : SUCCESS in 1h 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7de764e546e95452773b9e29b727e745cebb4ccd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-01 05:40:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9196fb16a48c3d0fc2bb5a67e40de6886ccc0a8a","author":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"date":"2020-04-01 05:58:29.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e028da9957e6d86e38dd27e9c8174812dfd1017f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 08:07:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20e61167d13e45488171888f49421a77 : FAILURE in 7m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2a878e3c9d4cd3bfff923127a31ee8 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b7b79e8d624df1aa39869d4d92691b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f70c3203ec404420885e107c4fb93b12 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1501285f71974a93ae16953612ead120 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7056a05c1a22467c946eb8c158d6bf51 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461471372376445eb37f7da2418df097 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6cd4a4386104d27b4019efdd01b1e49 : SUCCESS in 10m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4986008c62a84db8bba4efa999f42936 : SUCCESS in 7m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b91144526c44834a6397fe4de7a543f : SUCCESS in 54m 15s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5ed5861c62e5439a875d9ef55c07a88b : FAILURE in 58m 26s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4bcd7ff56e894b90af5c6a2d5acf66dc : SUCCESS in 2h 07m 04s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0fd712c1f4d24044a75ae2b994a6ee44 : SUCCESS in 1h 48m 37s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/db341f256b774342ba9cc443c5ec6ec5 : SUCCESS in 2h 06m 03s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/747ad46d40e1462f8f3a7898b77b12ff : FAILURE in 46m 35s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/902f41dd5585456e9edecb70e4a60e8c : FAILURE in 52m 22s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ed9b25357c424ee087743cfce7c0e1de : FAILURE in 1h 06m 24s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5671fc238bd431c8b21c24c55a1650b : FAILURE in 1h 40m 06s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/99aed27d3afa48269b7de60ba3d1024e : SUCCESS in 1h 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0820977d91a7e94886a75dbfe5b52b91ea82253a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-01 10:52:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"352430f9a07e3f44a9912e681a1de680c1196551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:36:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/333889f8cf274dd2804fd05ef3f2f972 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef7e26a72a9e41ef9fd0eef6d5f782a7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb864dff0bf549a4bce953f5fd6b984a : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7591ccdef684ed0a3c3b1c0e450b4c0 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/baedcccd84584ecfa8a95c658a804470 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c00b04d204f43d09440d894b19f81e9 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7aef47d42414c0d9d78b1023e71e2bf : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/253c302101dc4a45b51fb41d33f575ec : SUCCESS in 8m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a6c16c85b444b4f9f3507dd182b438e : SUCCESS in 6m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/28776e6dc28f4005b949c8df16dd2244 : SUCCESS in 59m 03s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/336f163762de41ecb9997a4e43fb0cfc : SUCCESS in 2h 26m 14s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/12012c4907bf4ac1abd50d21dd0ee09a : SUCCESS in 2h 43m 43s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f0d5b472b9e0444583f96994354a04d0 : SUCCESS in 1h 33m 03s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b7ee799c8b764ae68feab2418603b730 : SUCCESS in 2h 15m 40s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7d6a1f2e9a4c47ba9c490fd470f3f37c : FAILURE in 1h 02m 19s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4b442f15bb77445d884b2b47cada0f62 : FAILURE in 1h 01m 05s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/69efab3fa0da4638855bcd6075509d2d : FAILURE in 1h 24m 38s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31787ac96bb04890b1addfbff18d40a4 : SUCCESS in 1h 03m 06s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/5a39626bd05b40018f477d952f007927 : FAILURE in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74e0dfb3fa49c812186b031d90f30a4a03f01059","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-01 20:18:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"36efe7a2aeb1195fc4ef14fb8c2755a349e62616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:18:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01988b4d84a0d949a991223852d6876e58a6e7b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:17:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"30d0e2bf25cb459419424c3fc0333b76d844d6a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:17:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c15c72d31dd84d0f9c87be4b2d3b19b1 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af70242ffeb34e1c95124101a6aeedef : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c83be39e238435f87295678ce80ef5a : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d377051ca72411d88a8cf29e27faffb : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e4b49074a9443a9b9bad14106e2557 : SUCCESS in 6m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2f2f186123944be8b4ac57981a6042c : SUCCESS in 9m 00s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/56bcc0faed494ad5837faf364fe6039f : SUCCESS in 1h 57m 52s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ac7beb8f0d9a458190108124d03a4a26 : SUCCESS in 2h 53m 14s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f615cce5954e42ae819d56a10828b612 : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":3},{"id":"6f5ac59101f647e79cc6bba49bc44973ba0bc887","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:19:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81c3b5b3c8c4b22a2c0a703183c0127 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d8fb455d5d7111424091800ebfb6147cfde2871","revisions":{"af496108a7b1d6eb19210d60f90773e8cb4c6e14":{"kind":"REWORK","_number":1,"created":"2020-04-01 01:08:24.000000000","uploader":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"ref":"refs/changes/62/716462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/716462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/716462/1"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 01:07:44.000000000","tz":780},"committer":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 01:07:44.000000000","tz":780},"subject":"Ass innodb configuration parameters log_file_size and flush_method","message":"Ass innodb configuration parameters log_file_size and flush_method\n\nAdd these to the validation rules for Mysql and Mariadb, as we are\nusing versions for which it is possible to change these w/o special\nprocedures,\n\nChange-Id: I9014f4ae3083173b49d0ee581c976d06e852a29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af496108a7b1d6eb19210d60f90773e8cb4c6e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af496108a7b1d6eb19210d60f90773e8cb4c6e14"}]},"branch":"refs/heads/master"},"74438f799386548383e909fd24efa82ab12a86b3":{"kind":"REWORK","_number":2,"created":"2020-04-01 01:18:21.000000000","uploader":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"ref":"refs/changes/62/716462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/716462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/716462/2"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 01:07:44.000000000","tz":780},"committer":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 01:20:39.000000000","tz":780},"subject":"Ass innodb configuration parameters log_file_size and flush_method","message":"Ass innodb configuration parameters log_file_size and flush_method\n\nAdd these to the validation rules for Mysql and Mariadb, as we are\nusing versions for which it is possible to change these w/o special\nprocedures.\n\nChange-Id: I9014f4ae3083173b49d0ee581c976d06e852a29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/74438f799386548383e909fd24efa82ab12a86b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/74438f799386548383e909fd24efa82ab12a86b3"}]},"branch":"refs/heads/master"},"9d8fb455d5d7111424091800ebfb6147cfde2871":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-01 05:58:29.000000000","uploader":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"ref":"refs/changes/62/716462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/716462/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/716462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/716462/3"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 01:07:44.000000000","tz":780},"committer":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2020-04-01 06:00:58.000000000","tz":780},"subject":"Add innodb configuration parameters log_file_size and flush_method","message":"Add innodb configuration parameters log_file_size and flush_method\n\nAdd these to the validation rules for Mysql and Mariadb, as we are\nusing versions for which it is possible to change these w/o special\nprocedures.\n\nChange-Id: I9014f4ae3083173b49d0ee581c976d06e852a29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d8fb455d5d7111424091800ebfb6147cfde2871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d8fb455d5d7111424091800ebfb6147cfde2871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
