)]}'
{"id":"openstack%2Ftrove~716725","triplet_id":"openstack%2Ftrove~master~I6e6638c07b5d24c90e9deae277e98149f383eb3d","project":"openstack/trove","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I6e6638c07b5d24c90e9deae277e98149f383eb3d","subject":"Update hacking for Python3","status":"MERGED","created":"2020-04-01 18:48:11.000000000","updated":"2020-04-04 03:43:06.000000000","submitted":"2020-04-04 03:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716725-1585971693174-40a9e122","meta_rev_id":"b0e92481ccba058d5504ca7ecfaee8b7bc20181c","_number":716725,"virtual_id_number":716725,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 03:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-04-03 22:55:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-04-03 22:55:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 17:07:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-03 22:55:56.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-04-04 03:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"261f0fde1f8f978448e46ada96fd7e9f6e6cb783","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-01 18:48:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"897b2b6aa274cab1dfef845635cd92c19f6c3d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 21:00:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5f27a31a22564cf99c3a65e0bc5d2b89 : SUCCESS in 5m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f5dcb0cafbb4cafbaebaeeea6c59af4 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df323f16c58547fa9acad0a0d7a5ce7b : FAILURE in 7m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a293dd1d04a84ac2b6c3f9139a3a2db9 : FAILURE in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e399d08808947e5a8d101e60ce89fa9 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2d4cff039aa4acc80eb3c075a877ad1 : FAILURE in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5fb45c67db245bba8195784efedb0f8 : FAILURE in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f6ed1bca3204169969b9c92a3f815be : FAILURE in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bcc3c1c64b43929c6a12f9f6045e9a : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/150764be6d0240caaa62820055c66d99 : SUCCESS in 8m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/032c2d3fe9bd4fcfab0eecca12600354 : SUCCESS in 8m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ca46393309d43dc9b171b46c97981b1 : SUCCESS in 6m 47s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/19750e95ce14487892875e00a609bee4 : SUCCESS in 56m 14s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b2ccaba6f9974cc3bea0bee3918e0e43 : SUCCESS in 1h 58m 03s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f8d7de69d6e8415aa1648d7a538d9b1e : SUCCESS in 2h 10m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0fea74f4d1dd4cd88ba291daedb05d23 : SUCCESS in 1h 39m 30s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e2a25a3f6bf7445ea0f6d853a5941771 : SUCCESS in 2h 10m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d7b5a927bf6c4f6e9fa6decdb99dc7ac : FAILURE in 49m 55s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/02ecc52bf3194a8abba97b58cdbb787d : FAILURE in 43m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e59b9e7ca10341589c501109bba748d0 : FAILURE in 1h 15m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3a2d530c47b4ab0bc9bf6358f88f4fd : FAILURE in 1h 38m 01s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/fc1ccfa2538c47acb0d7baee34eca804 : FAILURE in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05221cd3d1e695e0bf0cd566b94b062d8376c0a8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-03 12:55:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"807ba77165970d660e65ad04c003de93b4be591e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-03 12:58:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68267bccb5fd4b637efd662d218fecccd7458ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:06:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64a8e30df5e34904aea1ef654f99615c : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2450b03aa54d404a9e3eee3640354c7f : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0b1b61c9c274242833a22ad2fff31e4 : FAILURE in 9m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a55af7623a74b50abbc63f9216b1446 : FAILURE in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848d794789cb496c917b70449d7a450f : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5204f10cfe34dd2baaee1851e74fa98 : FAILURE in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19cd05dddb8c432789a600c160420802 : FAILURE in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5ecfd5c012d4b0f9a592d451eca1328 : FAILURE in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0491ed977a494d5799e2113ea796b6fb : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7711caf6cd1416187d8398e068d3590 : SUCCESS in 9m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a5e341d4d4a4648b57c864b8d57aac2 : SUCCESS in 10m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b87de9c61564ec5ba0dada98b75485f : SUCCESS in 6m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/398fb91c795d45988085e28139f828d5 : SUCCESS in 54m 21s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3d281a52a84e42df945b83f603099509 : SUCCESS in 2h 08m 25s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0030bb7563ae429b86967313cb606912 : TIMED_OUT in 3h 05m 16s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4dfeed6c18b940e49720f5de316ea77d : SUCCESS in 1h 39m 17s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/8484faf812fc4dee871d00a833899710 : SUCCESS in 2h 16m 10s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9dc463b4bf8b422e97fe4525f564cacf : FAILURE in 46m 01s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8471beaa63d940a697727f55fb1d7793 : FAILURE in 42m 46s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/65e714bf922b4997b2a99f7746987714 : FAILURE in 1h 05m 55s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/437f3483d7794323beefce91989efbb8 : FAILURE in 1h 30m 13s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/4c77efa6dbc746b29ad469a0c6ea3cef : SUCCESS in 1h 57m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e84c938a8eac66d875121b397247e49df303fa4e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-03 17:07:54.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f63d5951f0a5c5a6afbb65fa647d2f87e548a60e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 19:48:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0cee47ca61b747a9b27508401333a131 : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f979c79ffc246808e8e175e82395f92 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b425270221aa40c1be90a3fe9c2b6ed7 : FAILURE in 11m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd996d18324c421db86029c37456cd58 : FAILURE in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f9cbd9940f410c99c0c7cb0158a606 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16001ed2b780416aae2b32f2d73c93e3 : FAILURE in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83d27bc5b2c64fc68e3212c5209f5982 : FAILURE in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cead090e7b154da486e2b3a1f6f28030 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4ca56102bc4e8cb868214d10862929 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e0737892e3f4b218cef817fe8372fe0 : SUCCESS in 10m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9022835c63be4044b9ac8ecb1877837c : SUCCESS in 12m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/514077a94d35481c91536220ea4764db : SUCCESS in 9m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a4c07f93a48e4b16a8a4da914b533b3b : SUCCESS in 1h 08m 27s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/93c9610bd18d4de09f48b0efabad23f1 : SUCCESS in 2h 09m 10s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/85051be66e464d5fa61541bab4b7486b : SUCCESS in 2h 16m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/29c25d29e9f1414ea6002f704ad6dc02 : SUCCESS in 1h 41m 46s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/dca0567637c74ca18b40ed943913e4ad : SUCCESS in 2h 00m 14s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/bb83f151e923455b82cd87f9cbf9f6d8 : FAILURE in 44m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5c33191af946434184eb8805c4c74f43 : FAILURE in 46m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dd59173b54aa4e12a99cf325d6490b7f : FAILURE in 1h 04m 28s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a2e2f6e73bb4cb9afba7cfa809af180 : FAILURE in 1h 20m 22s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/7ff08c1f6cfb473a904798bf6fe9288d : SUCCESS in 2h 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71117bd45ac7e6473c5bd62428074f423a2c3ff2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-03 19:51:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ceb43629d46af6f96bcf7eeab8c9f4e08e430d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 22:24:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1f7ee5b0bbd94368a7e0f812fff134c2 : SUCCESS in 7m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6fbbd608252546d4aaa725966e91d808 : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8c69fd8a35c4a30a83a54c7a9e305c0 : FAILURE in 8m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/066a81b1ac964b61b0514158357041b8 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7380e6332c5c4f1fa2619157d97c4e5b : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32dbb02e5b9f450bb34fcf1f92d2e0bf : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31a56365bc2d4a048b9e3c8b3915a155 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603cc4a319924b07afd4b00bf620be4a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4664adca5d5d4bd984d12b2d509a0c57 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9f68c088784e108b29558e32a2b37a : SUCCESS in 9m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba35dac2df1b4e0881e672154c2fedf3 : SUCCESS in 11m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd7c14a32fde44f693762fcb49480bfa : SUCCESS in 7m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9d1ed3e29fd4451c8790eea638ca4862 : SUCCESS in 58m 02s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b51c9541977b46058e9a2382e2b2bbfc : SUCCESS in 2h 16m 40s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c6b796deb4bf4b2aa89952617f87a14b : FAILURE in 2h 25m 42s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d773d4434fa846ea89b8b4d29ebcf7ce : FAILURE in 1h 44m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/93ce8cc4171f4abda577662cdb5931aa : FAILURE in 1h 05m 26s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9b1097b0bc03417dbe4d32787ab96c6c : FAILURE in 47m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8f387a69f47b421d9919155d2a83a3b3 : FAILURE in 45m 01s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ad854b4b4092486a9fc8f6b6e37b4ae3 : FAILURE in 1h 03m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8947645f33dd4c34a49b43f9f7cce10c : FAILURE in 1h 10m 29s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/9d992a5f3b474a4c8506170a944f174a : SUCCESS in 1h 54m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36b57251f4b0d95afcafac75bcce331240be72a7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-03 22:55:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e5946e26ab404fb954bb874349ddefa85eb52a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 01:34:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4410d189eb714e4b88169aa67765dbff : SUCCESS in 5m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2cc45982f66f4f918af9f4df5d4b2c7f : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84236aeb15cd4e8085ceac2a7947b3a0 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b3274132fc45d5b69ebd17217fe175 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bee9f3b55184c18a1cc8ef7b4628e5c : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c58d283352ec4fa081d3c96bd099f3a1 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06eefd40c08e4010ab06e0e5abafa4a6 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a391393111cb463482290654d0f077fa : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4114ad1776594adb856f487e2de471c7 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8346e615567b4a34bb941382fb3e390c : SUCCESS in 6m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2f77e03a1a543fa876422ffd73be5f3 : SUCCESS in 8m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30c209b00fda4eb49c8250176e72b416 : SUCCESS in 5m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c795dd38ec564523afa42e8f887ac7dd : SUCCESS in 1h 01m 57s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/70374af1fa534289a156eca886893ad0 : SUCCESS in 2h 00m 12s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/71b3a6ff4fb244f0a0c0824a9a85dc8e : SUCCESS in 2h 33m 02s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6124c7f2642248ebb57300961480afd2 : SUCCESS in 1h 48m 44s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a0f2086233474d69b41983608ccaedcf : SUCCESS in 2h 22m 20s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/cd711200f5d841a28375657181787cc7 : FAILURE in 44m 14s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4898bd84d01e477e86628e1bb9b258c5 : FAILURE in 41m 10s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e4693f5de70c4a40aa88842f1c1f705d : FAILURE in 52m 52s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d81d4edbb2334b9a85fd2619b43371dc : FAILURE in 1h 25m 25s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/ee9c775631e44166815d7154cc5fa74a : SUCCESS in 1h 54m 27s (non-voting)\n\nWarning:\n  Job build-openstack-api-ref: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/api-ref/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_messaging/rpc/server.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/eventlet/greenthread.py in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/cover/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job trove-tox-bandit-baseline: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/bandit/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/pep8/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/api-ref/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/eventlet/greenthread.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file trove/tests/unittests/extensions/common/test_service.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file trove/tests/unittests/common/test_wsgi.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file trove/common/context.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file trove/common/stream_codecs.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file trove/common/notification.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/bandit/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file trove/common/apischema.py\n  Comments left for invalid file trove/tests/unittests/common/test_context.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file trove/db/sqlalchemy/session.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py38/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/cover/lib/python3.6/site-packages/oslo_messaging/rpc/server.py","accounts_in_message":[],"_revision_number":3},{"id":"940b466a61eed159c93caf10b52fd1ff8e681c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 01:34:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"021c9aef7b6d25a10b20b308d24d651c336cd58f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:41:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bee38f1282334c5287a9fcc91a137547 : SUCCESS in 5m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/049703e184ad482abba87bfdcfb7c549 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfbf95db0b4c41ed8cef725c8abfc6ca : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b91ae5738c4b079ca3fdbc14e69c9c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b427085f758f456bb64174e7b599a03a : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed03410d413b449f9cc3087b0fc44800 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308a8ba8fa82459182e4ebda6ec725f0 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a07133b889f4b45b8b37729f2433098 : SUCCESS in 7m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09766798134f4c93b69d0d6165533d66 : SUCCESS in 7m 50s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/702d59042122461e889376152425abb8 : SUCCESS in 2h 03m 37s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/19bd31bd66aa48659fb6be879a218370 : SUCCESS in 1h 47m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/694314a90618463ca4aaee357295cee7 : SUCCESS in 1h 24m 21s\n\nWarning:\n  Job build-openstack-api-ref: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/api-ref/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/nose/importer.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/impl.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Comments left for invalid file .tox/pep8/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/api-ref/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file trove/common/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file trove/common/apischema.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file trove/tests/unittests/common/test_context.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file trove/tests/unittests/common/test_wsgi.py\n  Comments left for invalid file trove/common/notification.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/sqlalchemy/pool/base.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/pool/impl.py\n  Comments left for invalid file trove/db/sqlalchemy/session.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file trove/tests/unittests/extensions/common/test_service.py\n  Comments left for invalid file trove/common/stream_codecs.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/nose/importer.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/eventlet/green/OpenSSL/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/webob/acceptparse.py","accounts_in_message":[],"_revision_number":3},{"id":"c30f6e9991db1fa9af6f6751abe65b4966e15064","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:41:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"487aaf5e75c4347cbcbe662f70b9b5b2a96beb26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:43:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9fc067cf14744187a736141b7ed890ee : SUCCESS in 1m 13s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6398afdade6d49a09f1aed1ea477d466 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce695b7381af419fb6d5c65ee942a998 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f33c2130511f09a6014f45c12345aa158ab9fb6d","revisions":{"9abfa268838c2c29ed018c3c8a6118d41cb142e0":{"kind":"REWORK","_number":1,"created":"2020-04-01 18:48:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/716725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/716725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/716725/1"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:42:52.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:46:47.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not\nneeded in installations.\n\nChange-Id: I6e6638c07b5d24c90e9deae277e98149f383eb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9abfa268838c2c29ed018c3c8a6118d41cb142e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9abfa268838c2c29ed018c3c8a6118d41cb142e0"}]},"branch":"refs/heads/master"},"01bc62a26f76a1d769fdfbfb266879fea12b53be":{"kind":"REWORK","_number":2,"created":"2020-04-03 12:58:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/716725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/716725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/716725/2"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:42:52.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-03 12:58:21.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not\nneeded in installations.\n\nChange-Id: I6e6638c07b5d24c90e9deae277e98149f383eb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01bc62a26f76a1d769fdfbfb266879fea12b53be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01bc62a26f76a1d769fdfbfb266879fea12b53be"}]},"branch":"refs/heads/master"},"f33c2130511f09a6014f45c12345aa158ab9fb6d":{"kind":"REWORK","_number":3,"created":"2020-04-03 19:51:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/716725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/716725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/716725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/716725/3"}}},"commit":{"parents":[{"commit":"3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e","subject":"Merge \"Devstack should install the trove-dashboad module by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cdcfac30b603ad4461bbcdde1b4cbcf3aadf49e"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:42:52.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-03 19:51:00.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not\nneeded in installations.\n\nChange-Id: I6e6638c07b5d24c90e9deae277e98149f383eb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f33c2130511f09a6014f45c12345aa158ab9fb6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f33c2130511f09a6014f45c12345aa158ab9fb6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
