)]}'
{"id":"openstack%2Ftrove~720878","triplet_id":"openstack%2Ftrove~master~I84e4a31f14f9327cc01ff2d699167d91112e1565","project":"openstack/trove","branch":"master","topic":"fix-2007563","hashtags":[],"change_id":"I84e4a31f14f9327cc01ff2d699167d91112e1565","subject":"Added checks for deleting datastore version","status":"MERGED","created":"2020-04-18 10:15:50.000000000","updated":"2020-04-24 03:09:01.000000000","submitted":"2020-04-24 03:07:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720878-1587697653057-276f548f","meta_rev_id":"108f1a5b23306c16ad53127aabda2bbbcb0756e3","_number":720878,"virtual_id_number":720878,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 03:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-04-24 00:50:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-04-24 00:50:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 03:07:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f895413390f6c79b6387d8018af34e0a255b609","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-18 10:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2651d6ebf98633e76f4a68e1ac61fbe6b17cd9d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 13:11:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc316cb4a4143ec9a07d492845b830c : FAILURE in 6m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d68c37eabd094eb28f6dd46beb6daaf1 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eaeebdb649b410c84a6f37dfcd9ec53 : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91fa295a91ec417d8bfce5b868c0c4ef : SUCCESS in 12m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc1d977013dc474aac8508c2971f522e : SUCCESS in 13m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb79caf45b784a8b8c0f30c8bc2c156d : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c67cd8cf6c49b9b33eea02803f2ae5 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e75cae80b84f9bbb54dd9a6e408eed : SUCCESS in 9m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a38dbf3d0d24b71a05863ae56489d73 : FAILURE in 12m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7e957f39ecc43d1b2f935158e5a911f : SUCCESS in 9m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dd94e376ab6e430796ecef265b880d53 : SUCCESS in 1h 20m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1fd881c068924de590b288490a5ac086 : FAILURE in 38m 42s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4662d15d50124b7a8658512a6f807eee : SUCCESS in 2h 53m 12s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7ce86400013648a7a2f147db6239382e : SUCCESS in 1h 33m 15s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/eedea328d2974ca8afccff7ae336e001 : FAILURE in 2h 29m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4ef8224cdb3440dbba3dc260e6764f71 : FAILURE in 1h 13m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8b9c9b13e3b94697aafe139014685439 : FAILURE in 1h 06m 10s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f79250c07ecb4a41b3be9b90db290248 : FAILURE in 1h 51m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b283a8d966ee4dcbb0eba83a09b855fb : FAILURE in 1h 42m 28s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/511f9639e339481aa6f1218d34929547 : FAILURE in 1h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10f0f6c648b578f063456ab711b33b0835f96501","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-19 09:00:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbfc75a2590dc36b18846efa0ee19033abd5dbc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 11:45:47.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/720f0926b49c41c9aee97fb08d1beed8 : FAILURE in 7m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3220187e47c4d79b57965115a56aa92 : FAILURE in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47b28b34c094dd6ac43a66ed46e36ec : FAILURE in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4a444b159144b1db7b01f9ee4706b7e : FAILURE in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65dc73113ac4464dbe59a2495855ca3c : FAILURE in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d189317db4974762896caa0379bbec5d : FAILURE in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c65e69d6324c45b3b7700b9ad967b8 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa2440a699224c5a8103c365ab56bca0 : SUCCESS in 7m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cc2ab048843439794764702575281ac : SUCCESS in 9m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/818f0912dea9405abad9533068f3a097 : SUCCESS in 6m 01s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a04c9595e13e41cf996f934a8bbe4400 : SUCCESS in 1h 03m 21s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2bae196a582c45b6baea6879b6296e38 : SUCCESS in 2h 22m 51s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e0b4faec8b2041d1b0cfdcf1c9c96954 : FAILURE in 2h 43m 58s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b663ab510a164fdc90da47c28eaf535c : SUCCESS in 1h 49m 20s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/68b7bff811ea44bea73afad92c1607e7 : SUCCESS in 2h 20m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/321f04570f7240109591f5d069fe01d9 : FAILURE in 45m 31s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6416c7240f674d179bb5a4f03f358fb6 : FAILURE in 44m 02s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/914dac9cb779490a8731a0b107d47b6b : FAILURE in 1h 10m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83557018249c4630969c6a5ba80ed07a : SUCCESS in 1h 07m 17s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/6484aeb3ab0a483eb94a440d0cc8eea0 : SUCCESS in 2h 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"121a9b8c33c6f111599a1bfc408a71483fe012a6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-19 22:37:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"522de171b47fa3fd3ed751e424f0ac35e92b41a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 00:56:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d9b389efaa143228e6ab7476e3cbe69 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/843aeb03e4294cac9379da4e3c2d287f : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd60f8164a5147c9bf2a2097a2958bd5 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38fd89621e544749cbf7a7e8bf9d070 : FAILURE in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5028506c080c44e698bc7ed3e1aa2e25 : FAILURE in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b308707377e84ad69e4f2b1e7132bcce : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf6f299a80245929607d0918753fb70 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b16831db7a104bfd85302efea8bb1681 : SUCCESS in 6m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4be7df2dafef4f23bab1518d337e07aa : SUCCESS in 7m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b4292f0c9b4483699d0ce97cb8abb99 : SUCCESS in 5m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1f170999f7444a88820aa26955577ebc : FAILURE in 1h 11m 07s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5f3a6adbf1db4d529e10a4ae6b5b30a9 : FAILURE in 31m 53s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/56419f11c9f147d4af04dddb45f4cc23 : SUCCESS in 2h 12m 06s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/736515fb75e0406eb0c9e5454c7b12cf : SUCCESS in 1h 39m 03s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/86b0c88ec7b144cb95b80e5637b3e0f2 : SUCCESS in 2h 14m 02s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d1c723d839884440b3720270f5b0bafb : FAILURE in 41m 32s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/1577412cddf34b12b52e10f714905175 : FAILURE in 44m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ba342d3ab97740ddb24998923cab1633 : FAILURE in 1h 04m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0ce5d361d5646b7b590d208d878308f : FAILURE in 1h 12m 24s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/678378eb8a7641a492d86d09ad6c3456 : SUCCESS in 2h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a2c7db9d33dc6dab84f4b2f3ae025527f2c32c6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-22 22:18:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9b44f21c87e5915c5035801c335f38ddb9f8a197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:11:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59b6106ae78c4253a38b7a20b8bcdfc5 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8097d4b79d9441fe9a569caec855e889 : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7645536abb2e49d58faa874fda895c9b : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/769b1a2f81f444ceb4fc1e91971cfa90 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0767aeea77e49669614d01d66276ea2 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7848bc3a6684d19bf5fdadc1cf7541d : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976bc76388024c69b4612c061cce2e23 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39d7d9238dcb40b492f07400a41dd7c2 : SUCCESS in 10m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6aa5d1af393b45c1bbea78496b3e17ad : SUCCESS in 11m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a20149feca64b4e8c49bdebc93012ad : SUCCESS in 8m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ded821d9944c40c2bcef1a89db676fbe : SUCCESS in 1h 01m 07s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f3ed25ad1ca541ea9f606717623898ad : SUCCESS in 2h 01m 21s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/49b73ca7b79a4016a3749b2e4f6d4c3b : SUCCESS in 2h 51m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/afb08d54c2f14d759e007f1aaf61693e : SUCCESS in 1h 45m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/97282bcd84eb4a17925afd3ee6284ca3 : SUCCESS in 2h 28m 08s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1d482102b6d34c2881b5db1afce42883 : FAILURE in 44m 08s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/a6495eaa307b477bbdf12b75fdc59f54 : FAILURE in 45m 36s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/32db6f1225e542d69fe3e9003abed2c1 : FAILURE in 1h 03m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d87b5704f154bdda3f61d736abe9c32 : SUCCESS in 58m 58s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/c34b293d0121468f86004c0aaab6a024 : SUCCESS in 2h 07m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5c7d8fd71f754724192a5a0f285fd4f30fec1fe1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-24 00:50:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"83f8ca72d1ce9cc50975a9b3ce5a914cd4b35186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:51:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91932d7fc4be0ba7f681340cf2474154800755a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 03:07:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8b1084755404a538228773ba13e7bca : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660b8b729f234b00b7e3ad5eef3ad6d8 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f8a997861349f4b1333d574eac3676 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e63b0eddf25495fba40087d97259a72 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417d19f76e6743b3bd81626b88bd46bf : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5b4c284a61f439e8c1720d6c16c7532 : SUCCESS in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/807b2039ed204d5e8abddfe25c20e468 : SUCCESS in 8m 23s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/336c284ff5634ab3b86eac4123427c97 : SUCCESS in 2h 03m 33s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8e62ad0b1a8e41e5bf4004e2625f4e79 : SUCCESS in 2h 14m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1e971f9c0cf340be99d8c75c5ed94923 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":4},{"id":"1498d93baba83c5e404f5e13481b28ff0e8848cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 03:07:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f97ca23f973e1523cff745950068edadf0771188","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 03:09:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef0e40d54bb4727a0062066350998d5 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91631aad80784866b899cafeb8ec0b97 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4057b10af44e7f21243e539dd2daad2ad61657b","revisions":{"c994f2cd4e87b6b1fabf9e16b3feb9f4e3551de8":{"kind":"REWORK","_number":1,"created":"2020-04-18 10:15:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/78/720878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/720878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/720878/1"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-17 03:14:54.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-18 10:15:24.000000000","tz":720},"subject":"Added checks for deleting datastore version","message":"Added checks for deleting datastore version\n\n* Hard delete the datastore_configuration_parameters table record.\n* Make \u0027datastore_version_id\u0027 nullable for \u0027instances\u0027 table.\n* Check if the datastore version is still being used before removal.\n\nStory: 2007563\nTask: 39451\nChange-Id: I84e4a31f14f9327cc01ff2d699167d91112e1565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c994f2cd4e87b6b1fabf9e16b3feb9f4e3551de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c994f2cd4e87b6b1fabf9e16b3feb9f4e3551de8"}]},"branch":"refs/heads/master"},"41767d692309163bf576ea36a28f39fb54e3549d":{"kind":"REWORK","_number":2,"created":"2020-04-19 09:00:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/78/720878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/720878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/720878/2"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-17 03:14:54.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-19 09:00:19.000000000","tz":720},"subject":"Added checks for deleting datastore version","message":"Added checks for deleting datastore version\n\n* Hard delete the datastore_configuration_parameters table record.\n* Make \u0027datastore_version_id\u0027 nullable for \u0027instances\u0027 table.\n* Check if the datastore version is still being used before removal.\n\nStory: 2007563\nTask: 39451\nChange-Id: I84e4a31f14f9327cc01ff2d699167d91112e1565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/41767d692309163bf576ea36a28f39fb54e3549d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/41767d692309163bf576ea36a28f39fb54e3549d"}]},"branch":"refs/heads/master"},"6b0195cd693d96dcd2352c48be096e2423747ac1":{"kind":"REWORK","_number":3,"created":"2020-04-19 22:37:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/78/720878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/720878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/720878/3"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-17 03:14:54.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-19 22:33:28.000000000","tz":720},"subject":"Added checks for deleting datastore version","message":"Added checks for deleting datastore version\n\n* Hard delete the datastore_configuration_parameters table record.\n* Make \u0027datastore_version_id\u0027 nullable for \u0027instances\u0027 table.\n* Check if the datastore version is still being used before removal.\n\nStory: 2007563\nTask: 39451\nChange-Id: I84e4a31f14f9327cc01ff2d699167d91112e1565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b0195cd693d96dcd2352c48be096e2423747ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b0195cd693d96dcd2352c48be096e2423747ac1"}]},"branch":"refs/heads/master"},"a4057b10af44e7f21243e539dd2daad2ad61657b":{"kind":"REWORK","_number":4,"created":"2020-04-22 22:18:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/78/720878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/720878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/720878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/720878/4"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-17 03:14:54.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-22 22:18:03.000000000","tz":720},"subject":"Added checks for deleting datastore version","message":"Added checks for deleting datastore version\n\n* Hard delete the datastore_configuration_parameters table record.\n* Make \u0027datastore_version_id\u0027 nullable for \u0027instances\u0027 table.\n* Check if the datastore version is still being used before removal.\n\nStory: 2007563\nTask: 39451\nChange-Id: I84e4a31f14f9327cc01ff2d699167d91112e1565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4057b10af44e7f21243e539dd2daad2ad61657b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4057b10af44e7f21243e539dd2daad2ad61657b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
