)]}'
{"id":"openstack%2Ftrove~721143","triplet_id":"openstack%2Ftrove~master~I7a281868b9579987fb9308badb9acab86d73f1bc","project":"openstack/trove","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I7a281868b9579987fb9308badb9acab86d73f1bc","subject":"Use unittest.mock instead of third party mock","status":"ABANDONED","created":"2020-04-20 02:57:50.000000000","updated":"2020-04-20 08:06:03.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3db0366e25035ea138da35f1c350a02b04908be","_number":721143,"virtual_id_number":721143,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 05:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3012de8fa90a172b3abbd55c6402b5a92288aa5a","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 02:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fbaa2d1cb7d53c0a62b4e4b01ae37c7c5cfb93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 05:24:46.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/926ee8ceea4548a0bd7df3ba602a0251 : SUCCESS in 5m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7759d46f571a482f91ec2b37379a09e1 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5fe99480dc84df797695b7296c1f41f : FAILURE in 7m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aef0fc539754b64aee80dad8d90ff33 : FAILURE in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7331e27dcfc46bba3d6e3e8679e40c8 : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b4187fbabb1482fa27d4bdeaf9aa21a : FAILURE in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc731978bffe45c0bff5b00ab77476fd : FAILURE in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7689a5772444b6e881daecf8a842e89 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcbe83391c048c9abef0795908b5039 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e476ba9ae5a948a1873b1207f5ef1629 : SUCCESS in 9m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db73d81610494d07982680ef4210c2e6 : SUCCESS in 8m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78300e8748974e21b0e3adfc79036baa : SUCCESS in 5m 47s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3accbb43d54f451dabefbcf78d8e66d2 : SUCCESS in 54m 35s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/20a106b5d12c44dfbde5fd28741d2c35 : SUCCESS in 1h 52m 51s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cd7cf1d733184f6ea958ce54659826a4 : SUCCESS in 1h 58m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/63d405e920cb47729aef5e3c6e65e701 : SUCCESS in 1h 35m 54s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/00dc19a9a6b04d7e8457f43e60f12926 : SUCCESS in 2h 16m 56s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9c1d8b7c634f44369c9f7234cd2f18b2 : FAILURE in 42m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/797d6410495d42bb8817d201dd16405b : FAILURE in 41m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7384968f7879436c9e1c0a247b91264d : FAILURE in 1h 05m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53aa28e3e27d40d0beb9c2e0eb3c06f1 : FAILURE in 1h 30m 15s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f4526938d494414fa04cd3140f77df37 : SUCCESS in 1h 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bb17b5734eef9d453bab0af697fa9cba7d3d2f5","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 06:52:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f5670e8a024f9fdb12403e9975ba8110e90f5a3","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 08:06:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53df10c2f7df719f74ae268075532ae8bbc8c41e","revisions":{"c59b7867d9bc9bbdee3bc8a31fe4c4cd1cc9a4be":{"kind":"REWORK","_number":1,"created":"2020-04-20 02:57:50.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/43/721143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/43/721143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/43/721143/1"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:23:15.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:57:47.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib\n\nChange-Id: I7a281868b9579987fb9308badb9acab86d73f1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c59b7867d9bc9bbdee3bc8a31fe4c4cd1cc9a4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c59b7867d9bc9bbdee3bc8a31fe4c4cd1cc9a4be"}]},"branch":"refs/heads/master"},"53df10c2f7df719f74ae268075532ae8bbc8c41e":{"kind":"REWORK","_number":2,"created":"2020-04-20 06:52:13.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/43/721143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/43/721143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/721143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/43/721143/2"}}},"commit":{"parents":[{"commit":"8c3df10aa53ca78f764c2580dc99793af03f88b9","subject":"Merge \"Change @property usage to function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c3df10aa53ca78f764c2580dc99793af03f88b9"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:23:15.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 06:51:33.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib\n\nChange-Id: I7a281868b9579987fb9308badb9acab86d73f1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53df10c2f7df719f74ae268075532ae8bbc8c41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53df10c2f7df719f74ae268075532ae8bbc8c41e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
