)]}'
{"id":"openstack%2Ftrove~723169","triplet_id":"openstack%2Ftrove~master~Ib0064d4ee9bf05f9c8e159482f5118756a1d3c09","project":"openstack/trove","branch":"master","topic":"remove-volume-types-api","hashtags":[],"change_id":"Ib0064d4ee9bf05f9c8e159482f5118756a1d3c09","subject":"Remove volume-types API","status":"MERGED","created":"2020-04-26 08:17:15.000000000","updated":"2020-04-26 14:13:38.000000000","submitted":"2020-04-26 14:12:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723169-1587910323859-593d93a0","meta_rev_id":"97a7f815de801bc1d2ab33dba795c846039cf5b3","_number":723169,"virtual_id_number":723169,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-26 14:12:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-04-26 11:35:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-04-26 11:35:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 14:12:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d59b16536611d29453faf4145c1321e1938ef411","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-26 08:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"957e63d45f45bd24e134de29fc6b8770448a393a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 11:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69689cdefdd541e1bc2317882d8a93e7 : FAILURE in 11m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e096b8bcd3eb4b4fb139780019d5af22 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f79a1d845574e15a5500589efb0efae : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77cd5643d5c54908b1acb369b05b1370 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6339014d069241008cfab40c5e8c2004 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5ed1c2893447ecb52a1d4cb4825d19 : SUCCESS in 4m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e11918752f82409496dff766175fa802 : SUCCESS in 10m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61d8a010384b42cba92f860366f4c8a8 : SUCCESS in 6m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7209fc689f05415184d227de618af7b0 : SUCCESS in 52m 46s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/19ad6d5404a94518a83222e9227b59a7 : SUCCESS in 1h 53m 14s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/0af4a21d7ba443c9ab05129c77db1fb5 : SUCCESS in 2h 49m 49s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/87d12bd721664ac9b1724c9bf303afea : SUCCESS in 1h 40m 58s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/9ee39a5ce6eb4f9d95c124242cb23063 : SUCCESS in 2h 14m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3eadcf4216d84d1f8232721457d7627d : FAILURE in 56m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/01a7d2f8e2984ff3a5d567ebb06fcc5f : FAILURE in 1h 37m 37s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/147b4d1823d74e0eaebe912a0d8210ce : FAILURE in 59m 08s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e8f4658d6584ef18489f300bf751d10 : FAILURE in 1h 42m 03s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/23bd422e1ec847438bdcb2de9a23b5af : SUCCESS in 2h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e92cc1a5f8b32d0f74d941aaeeafa1089a16aae","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-26 11:35:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81da6bea3006c12ea82d82c206c8ca3143eaf544","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 11:35:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60394e297e5d128a7e9eb56d3949a9a58d617487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 14:12:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aafadb83528041f483df10d852cef72c : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbfb4879ee4d4a26b8a43a9c2ef188c5 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf096f064524feb8c6e6968d2fa0962 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6d98622a0e742768a26127b7110793a : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2ca406f64045b7a624bd151e014018 : SUCCESS in 5m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/047d915518ef4d5d9446ec0682d6138c : SUCCESS in 8m 14s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e9d0352999dd43d0a623280f6d54cd5a : SUCCESS in 2h 05m 27s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6f91954d59e44f17b2681d8d390aeae2 : SUCCESS in 2h 31m 04s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/59579d4d88d34aac8f85b14c45d2d106 : SUCCESS in 1h 32m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6816d3ea7f47c07956463e4b4d606bf56c89e577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 14:12:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d8721e579fe0d619a7c8605d6a7515e8889a69b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 14:13:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feff7573b3184d5a99642305c3aea122 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"feb982547afe26be4e885b75d82fa5d51a784c0e","revisions":{"feb982547afe26be4e885b75d82fa5d51a784c0e":{"kind":"REWORK","_number":1,"created":"2020-04-26 08:17:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/69/723169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/723169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/723169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/723169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/723169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/723169/1"}}},"commit":{"parents":[{"commit":"06fadbf0543058ada92b5689ecf6c1c14f5f033b","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06fadbf0543058ada92b5689ecf6c1c14f5f033b"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-26 08:14:37.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-04-26 08:14:37.000000000","tz":720},"subject":"Remove volume-types API","message":"Remove volume-types API\n\nChange-Id: Ib0064d4ee9bf05f9c8e159482f5118756a1d3c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/feb982547afe26be4e885b75d82fa5d51a784c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/feb982547afe26be4e885b75d82fa5d51a784c0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
