)]}'
{"id":"openstack%2Ftrove~725672","triplet_id":"openstack%2Ftrove~master~I61e5e270bf66b0355da3282c19cbc9fd42c4090b","project":"openstack/trove","branch":"master","topic":"bug/1863021","hashtags":[],"change_id":"I61e5e270bf66b0355da3282c19cbc9fd42c4090b","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-05 19:42:46.000000000","updated":"2020-07-18 12:09:47.000000000","submitted":"2020-07-18 12:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725672-1595074119078-bdb66bb5","meta_rev_id":"1465b51e30ccea59993eff2f53a52821da6c1a99","_number":725672,"virtual_id_number":725672,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 12:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-07-18 11:28:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-07-18 11:28:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 11:28:16.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-07-18 12:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d71ad31d8c4d674c234f6079604fa06638139b99","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-05 19:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a188f827f53930a9cc3eb966ff3db9ddd7ab297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 22:03:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b4277889914394be6e36c81f1dd473 : FAILURE in 9m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f18f449136024db4ac4b1ae4f0b03017 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0d267a55274b9eadaea9a7310df844 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d911a23a5768435da683033e5f4bd9c4 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7f52d8c69e64f878995e1a3fe0eda75 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0354bcd7de0949fab579225b35b9a237 : SUCCESS in 6m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6af021d61dd44f7a8b3dc0c01e009a1 : FAILURE in 10m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce137eb012544d5fa6527a7f80f0c897 : SUCCESS in 6m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/057957ce8fda45ea8dda527a6ddf2cc9 : SUCCESS in 58m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/04f167b2dced4f8c8326f4b8ce9c0d1b : SUCCESS in 2h 00m 47s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d132d4aa181b4495a924f1f145542ac0 : SUCCESS in 2h 20m 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/dc67e868bd464a15b57a86a67aba3db1 : SUCCESS in 1h 33m 34s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/dfbd53f10ecd4c639a40aa43ccce20f1 : SUCCESS in 2h 08m 48s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/99fa27b5b27141df959c65501a655af3 : FAILURE in 1h 03m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6d6390b1302045138fd4e278d530dc70 : FAILURE in 1h 00m 32s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dca09041d5024d1eb9b5d2d7ba8c3e02 : FAILURE in 1h 06m 03s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29cc544d258a4db3836ce47c719eca0e : FAILURE in 1h 46m 54s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/eb02a0bbf00247d1b8bb24f4f9a266bf : SUCCESS in 1h 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe9b4f4120564b804845ef80a1d01e12eda7fb94","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-06 13:23:18.000000000","message":"Patch Set 1:\n\nerror doesn\u0027t appear to be related","accounts_in_message":[],"_revision_number":1},{"id":"3ed745c47b0b6cd1bcdef7cbc49de0c57a149018","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-06 13:23:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7ed44feb2aadcb0263904f8e8d94872cbf055655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 16:00:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d07f1bae02384b42b2609f7e57a9657c : FAILURE in 6m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f4bc3c3f5245a2b0d61b2883109577 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca3a427feb74ff384d3684bd48f125a : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00245428d89407a91221dc9cbcc6a10 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b011d5a0a0d64a6d8c64cb64a4487f3f : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095f03b82776435992f2d799695b3829 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5cdd7d3f09b64981ab193ac622eb7f59 : FAILURE in 10m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6fa569f547264afea5b0068781eea4bf : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/81d2074778c04d3cadc7cfa5c4ac78f3 : SUCCESS in 1h 07m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/461878ee41664078954144bc96518703 : SUCCESS in 2h 05m 59s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/95f8d996285546d4bcb95e09be4376b3 : SUCCESS in 2h 35m 02s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/10e202a087a048f6946f651b4a39b8b3 : FAILURE in 1h 36m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d0b0cef6bc404399bb8ddf4e723e1135 : FAILURE in 2h 19m 31s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1a2de3f75b2b4e618ed610e9b1c8b53a : FAILURE in 42m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/573067bf44104a48ab5ac2af74909804 : FAILURE in 1h 38m 18s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/afae77c50a114e07b160e2106be9aa93 : FAILURE in 1h 08m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47a89f2bcbf04303b97730bc895bf98e : SUCCESS in 50m 57s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f20f48d85339439e9fd7b1383c4ed975 : SUCCESS in 2h 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e255b3a975d8153aa97d3bb07e68e64dc6463171","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-07 11:25:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41e75edd06cd48f5a94071f884a35acf9445a98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:53:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0862e3dfa7b944cd8962ff3eea8d6a6b : FAILURE in 9m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c99ea5b551604c68a798776b2e91655d : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedc19f3553b4fd4ab16c41c67418f92 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e386ff1989044eb38e8ca24fdc5eebcc : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b92ed15254174197b267fb6ff42cc326 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e6d00363da42fb8c3e14a20eb3c131 : SUCCESS in 9m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a527587fdad1462d9ff080e590755e90 : FAILURE in 12m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f8c1aed0662446eaa51e476b1724dee : SUCCESS in 6m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b280d8678a4442c3b9d2570709f8aab9 : SUCCESS in 58m 30s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dacf4cdc4077403c956f65a62413dace : SUCCESS in 1h 58m 06s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4625d298a97e4efa9195b47a42070591 : SUCCESS in 2h 14m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7bf731826dc545c1b2fe9aaaf2fc9576 : SUCCESS in 1h 34m 47s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ab3624a0720c438fa3670ff488d93f88 : SUCCESS in 2h 16m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/33c2fd3d76644b699fbc3a80dc3406f0 : FAILURE in 52m 06s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3527b5a8d4f64c5a9f94170c731a07e4 : FAILURE in 56m 01s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/87cc80f3ea7a458e967167e04797195a : FAILURE in 1h 05m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0e82aa214db4e06895f450f83b35ba1 : SUCCESS in 1h 04m 21s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/3cd049710ddc4b7b950050d5840ec156 : SUCCESS in 2h 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57fef627d95d2c7a688d4c14cdcef5ae666e66b9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 01:57:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a6742c75ddd1fc0c050d9896cccffd6051bbeb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 04:16:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47b274f88b364372a6750e6ce68989e2 : FAILURE in 5m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27b1e4d90c984a4687ed72183b033404 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1efe4181a4d34f42bec598356f285866 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f51a92eebbf746a3a8010438a08c1d7d : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b9c6acfd8f2429f96094498e7a69bfb : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7d4b1f0f3444f498ea5bd1dc1228e7 : SUCCESS in 5m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d319ab8cfc04418b8bd5993c71f537c : FAILURE in 8m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ebde853d6134beabad7a711da2b6496 : SUCCESS in 5m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3097b1c702df4fc2b2d7a5878f25ba23 : SUCCESS in 54m 41s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b28d9902096c49539e319e292794cd5a : SUCCESS in 2h 01m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dbfbb43d1c2a4692ada9c900bd06c085 : SUCCESS in 2h 06m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7be8ddd98000438b83ee622a79b3a8ef : SUCCESS in 1h 30m 45s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6d19b19d70d24de18ab790132550115e : SUCCESS in 1h 51m 23s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3540ec6071854e18ad4f6d35a3410092 : FAILURE in 40m 56s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/231631e3f47b411491de7d9258885aff : FAILURE in 37m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5377bdfb9f2b46d2a7c56b3396bac82d : FAILURE in 55m 52s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20af63b75b4c4d16b73374c88b160758 : SUCCESS in 56m 44s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/4f057fd461a24470a99fe9ac882323fc : SUCCESS in 2h 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"931c095967833870e693d632c13c3d850641acaa","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 10:55:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f06e7da6bd6b6c8555ce4052da8d51d175d3a9d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-24 04:10:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"af40b34fa2553b3ebbec77a3e5b3de1208339ca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 06:14:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39071152dbd64d529ab8ba70a7d7516e : FAILURE in 5m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af406ccdf1cc48e1bab9f5e8cb77daa4 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67434c97315d41fcb8e4114f021aabfc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4d48faeb4224a3192a4f00de0215ddc : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c9b3a871b944e17a60d05112eb7b798 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a59de2cfb154b079432fa820a5e4bad : SUCCESS in 6m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6744c5900bc47779486b11b7a65a9c4 : FAILURE in 8m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b060dd0f03cd4cbba2426989592f8b11 : SUCCESS in 6m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/94c81fd8b9d54e94877a483a7ec0a56e : FAILURE in 2h 03m 07s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/601edb2b685441f5ba8df1e1756ff7ea : SUCCESS in 1h 04m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73f73f1d9d966138dcd78b4cd3146f81ac205a99","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-09 22:33:40.000000000","message":"Patch Set 3: Code-Review-1\n\nI don\u0027t think rebasing can solve the issue \"Unable to import \u0027__original_module_threading\u0027\"","accounts_in_message":[],"_revision_number":3},{"id":"99be9fc0158ac09d8cd5721cbbc5bf0ec3c1028e","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-07-16 14:03:59.000000000","message":"Patch Set 3:\n\nI think we can add a # noqa to that line like we did for other projects.","accounts_in_message":[],"_revision_number":3},{"id":"99c86cafb651883e49a0154aa583c3ff86236576","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-17 04:58:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"50cef04da292e3473a21de0f8ad1611e4c3742b7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-17 04:59:11.000000000","message":"Patch Set 4:\n\n\u003e I think we can add a # noqa to that line like we did for other\n \u003e projects.\n\ndid that for you.","accounts_in_message":[],"_revision_number":4},{"id":"2f36ca9d5e3d04d5f74e0e49f16abca82dd9b320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 06:56:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71c917972b34415bb0d963ac91abcff7 : FAILURE in 7m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8480be54c0415d9b48c738fbf816cf : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f91e6fa783749bf9a6c1df24ee18e39 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d1169132904bffb0cf9d9dea11245b : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e51f09ee6e54ac4aa5f578d4895871c : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cecf2ca741486b97b98ebc5de43cc1 : SUCCESS in 7m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb9362a2e4a24f998e28738ef7cbcf97 : FAILURE in 6m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/269a8b2bd4a549caa0f047319ce30724 : SUCCESS in 6m 55s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5923cab095294340abbba1036950202f : FAILURE in 1h 55m 55s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b0c0473b0a4479c9e18499ea64ff224 : SUCCESS in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca40a9014923ecd48aca418db1078209cda74459","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-18 10:09:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"62d9145c15c9875d88efeb36c7fd92f129e4c787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 10:43:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5dd1f4492b94622bd218026c8fb2d8c : FAILURE in 5m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d757a3237e054838804e6310fcdbb899 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2a188655c944939da4e6f116ab7ad7 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14b6a3178a67493a8cf72ea38ebd9289 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff3590b52d24e13b08bba767cc09566 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495e0dadfae14a8288bd30eabae5fa43 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06b9932ca7234f96ae0b8a891f7e175a : SUCCESS in 5m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ea180d8d9d614e7bb1a7f671d2f089d0 : FAILURE in 27m 19s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b3de45e19dc46118b94c241eb2dcf06 : FAILURE in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f2eee6171618e4c5cfd752b59db8a67ca5efcc5b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-18 11:28:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6f09f3a3bd43be94003d86635eafd89b88ca81d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 11:28:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6af88cbc746aa15c6e0ed5d5c91419afe9275cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:08:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2394caf378d430b8725ca19c92282f0 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e96499774b4fd0b1a8c819bcb4c8ac : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926c6d79d74643cebc764042ae87cd06 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f198e01b970142ecb8ff5e142deec994 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c59616775b447dcbe5b8a222ffe35d9 : SUCCESS in 6m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f18d8f931ff34788b1821178a9dbd3fb : FAILURE in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db311f6ac76f9a4378720d69534c560bca645c20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:08:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"22d59ec3d687d7dbda65303f1c08376d5d8c6824","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:09:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7c80fb8d7f475fb2730f2b048c06ec : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"56bb6f39b6c808160d2e4c96fcc2104ba1daf817","revisions":{"d538db7a3b689beb1242b5c2b774ce0d42478805":{"kind":"REWORK","_number":1,"created":"2020-05-05 19:42:46.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/72/725672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/725672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/725672/1"}}},"commit":{"parents":[{"commit":"8887503e25c8086259a86586105551497315f894","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8887503e25c8086259a86586105551497315f894"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:31.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:46.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I61e5e270bf66b0355da3282c19cbc9fd42c4090b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d538db7a3b689beb1242b5c2b774ce0d42478805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d538db7a3b689beb1242b5c2b774ce0d42478805"}]},"branch":"refs/heads/master"},"098aeff44ed8d9d0eadcad23fc52b830d2ee3342":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-27 01:57:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/725672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/725672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/725672/2"}}},"commit":{"parents":[{"commit":"523d66e8fd5d29be8dbae9aa79c5348d3dce8c64","subject":"Fix CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/523d66e8fd5d29be8dbae9aa79c5348d3dce8c64"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:31.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 01:57:58.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I61e5e270bf66b0355da3282c19cbc9fd42c4090b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/098aeff44ed8d9d0eadcad23fc52b830d2ee3342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/098aeff44ed8d9d0eadcad23fc52b830d2ee3342"}]},"branch":"refs/heads/master"},"6b59bdafbebe4eb617efce2c33276ecacf1441c7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-24 04:10:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/725672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/725672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/725672/3"}}},"commit":{"parents":[{"commit":"0520181a334e61b4f30e78b1459a766ce0294752","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0520181a334e61b4f30e78b1459a766ce0294752"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:31.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-24 04:10:32.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I61e5e270bf66b0355da3282c19cbc9fd42c4090b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b59bdafbebe4eb617efce2c33276ecacf1441c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b59bdafbebe4eb617efce2c33276ecacf1441c7"}]},"branch":"refs/heads/master"},"d255c73174c0f290a6b725925845c25c036d901b":{"kind":"REWORK","_number":4,"created":"2020-07-17 04:58:55.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/725672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/725672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/725672/4"}}},"commit":{"parents":[{"commit":"0520181a334e61b4f30e78b1459a766ce0294752","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0520181a334e61b4f30e78b1459a766ce0294752"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:31.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-17 04:58:51.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I61e5e270bf66b0355da3282c19cbc9fd42c4090b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d255c73174c0f290a6b725925845c25c036d901b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d255c73174c0f290a6b725925845c25c036d901b"}]},"branch":"refs/heads/master"},"56bb6f39b6c808160d2e4c96fcc2104ba1daf817":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-18 10:09:51.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/72/725672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/725672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/725672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/725672/5"}}},"commit":{"parents":[{"commit":"828e873846495df6d6911f9c196ae1fb478492f5","subject":"Support backup strategy API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828e873846495df6d6911f9c196ae1fb478492f5"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:42:31.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-18 10:09:51.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I61e5e270bf66b0355da3282c19cbc9fd42c4090b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56bb6f39b6c808160d2e4c96fcc2104ba1daf817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56bb6f39b6c808160d2e4c96fcc2104ba1daf817"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
