)]}'
{"id":"openstack%2Ftrove~727592","triplet_id":"openstack%2Ftrove~master~I6a80bd28d4bce6140ba5c3fabe7199301ac026cb","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I6a80bd28d4bce6140ba5c3fabe7199301ac026cb","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 02:24:15.000000000","updated":"2020-05-27 11:43:55.000000000","submitted":"2020-05-27 11:42:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727592-1590579722726-2a4b48c7","meta_rev_id":"45c0cb158a6e96a11321f056035c99c19c9e210e","_number":727592,"virtual_id_number":727592,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 11:42:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-27 08:29:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-27 08:29:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 08:29:48.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-27 11:42:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40a67bcaa736b5253a4304c8e626cf5b9375882d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 02:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f4772ac4e1250ce90f6de17431e4dbe2a8cc8a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 04:57:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bed2dc7222ae4f0d9436f1427ef6fc59 : SUCCESS in 6m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2bc5ad52fdac4d64bb99a82040ba1ebe : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0efa9832b24e4e05bcf5e39efd616bd5 : FAILURE in 7m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5bff13ea9204f4b8e9dad4c5ff489f8 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a0f0c03e624323b81da3c77e11eb17 : FAILURE in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d9aa14efd1a43ccaadbfb6f4b0eedc8 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7659fc60a99241dea46059bc39f9607f : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecc4e5552a34aeca667786ac9f15c5c : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa39bf2118314723baafa1981e04291a : SUCCESS in 7m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d76ca97e430a4fa281c279b24450849f : SUCCESS in 9m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/31db7be894cf4093aa43ec909aa88856 : FAILURE in 1h 08m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/10834c40aea940829f3924a83288ca9f : SUCCESS in 2h 03m 34s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/08aa8e79de2d4cbc95d8904dbe3792a1 : SUCCESS in 2h 26m 11s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/79392ae0b3f24d6bbb5f1495da470e4f : SUCCESS in 1h 57m 21s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/dfc0e03082f043308e868ba79b01359a : SUCCESS in 2h 04m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8810dce17f214346ac1526972672e65a : FAILURE in 48m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9358928ab00a41bd9d0d0074c16a1a66 : FAILURE in 41m 37s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/62ae35f7500345b0b166f255b89ea3f5 : FAILURE in 1h 24m 02s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab16966f9c54946a0dae1c99a2cb855 : SUCCESS in 1h 01m 15s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/57f604ca69b540fd81e6fd0dd4bfcf18 : FAILURE in 2h 24m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/instance/views.py\n  Comments left for invalid file trove/limits/views.py\n  Comments left for invalid file trove/tests/api/limits.py","accounts_in_message":[],"_revision_number":1},{"id":"21663e56799300a548c619ee1912b41aead010d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 01:57:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d4f0cb35736fa699a277fef52d1cf3f03db5fae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 04:38:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/15a9b2d67f994510b24e363a0a34bf8c : SUCCESS in 5m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bc52162050a43f088482249aa4a5cce : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd44c8cc58a4eada3c021ebe5849cb9 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/694a2a21f28f425ead981aabbcc85311 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e04842eeaa8f4a6f981d01db2e2db556 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc47cc5e9e6a484ebc62dd5c101410d7 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a63115cf65f4f62803a2345f688cc88 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07da41e3f50d4256ba737c8d5a7e2ad0 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f67cb741af0f4bc5a1b909c4bdca93f1 : SUCCESS in 5m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/389f58a04fbe470bbe9c1a209976899c : SUCCESS in 9m 12s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/72fa9dff0cb145b9bc7e9b1c40860c6d : SUCCESS in 53m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0e096ebf0aad4c968c7f5906ec4a3975 : SUCCESS in 2h 19m 12s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1cec07f03097402ba43fc26236f6b0f8 : SUCCESS in 2h 33m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/69283961785a41eab9a58747c98683d0 : SUCCESS in 1h 34m 40s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/369af023ddea4200ab7b6ee0396e9c3f : SUCCESS in 2h 13m 37s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a56d31cac83c4fe08d375bec61b9ee9c : FAILURE in 36m 33s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fef25d15932f40a4901619502dd01160 : FAILURE in 41m 26s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/a5867ec32b1f49f3a9f0084698064619 : FAILURE in 52m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4633a457fc4f4e18b7af102b7f5770cb : SUCCESS in 58m 12s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/947f332275ea4c6da0d8b1219a4a78f9 : SUCCESS in 1h 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e31c4d1fab07d70eed33754f3aaac46062dcf7d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 08:29:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8fea82149fbdd569559feef39a63f6d71fc54f19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 08:30:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c67d76ac4fb6fc8649607a7505ac299fa26ce31e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:42:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d6a41a111af448129572ff718911e7e4 : SUCCESS in 6m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9067cae44dd04c84b9230cf761ed5a18 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88ffeccc6cd54c63b185bdb2585bd8fa : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81bb41ee9214ecd980e2ecd8628b8ca : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4fa9552ab724b62b8fa2cc5e4b76bfb : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ec37eae50664d4d88673e815f08a4bd : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd670f8f8874297a1177ff80a2a9ce7 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/196ff3ff255b4ef8ab1119cb3f7dfd7a : SUCCESS in 7m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e50e746aea6438eba378eeb7c84aff3 : SUCCESS in 8m 28s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/51d4d6ee23e54d82bda309c332d8fad3 : TIMED_OUT in 3h 05m 23s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/62f314e66fae48efa6024ad1b665956f : TIMED_OUT in 3h 05m 32s (non-voting)\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4d5948b0818740f5abbab05a00c40540 : FAILURE in 2h 55m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"afb74f8d6a77986a671ed7a044274de10283aae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:42:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"45c0cb158a6e96a11321f056035c99c19c9e210e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:43:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6febbd69ade64f399b1a5d46206dc46e : SUCCESS in 1m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92c1c6e650448a7b3ec159d9b51015b : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e95bfdc3014681b4e870997a693cf5 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c5f5c90839a8f1fa7ad62841fe923a16317fca7","revisions":{"b159540c1f03d06c55da1e5fe01c36d214ddb373":{"kind":"REWORK","_number":1,"created":"2020-05-13 02:24:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/727592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/727592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/727592/1"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:23:48.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:24:14.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I6a80bd28d4bce6140ba5c3fabe7199301ac026cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b159540c1f03d06c55da1e5fe01c36d214ddb373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b159540c1f03d06c55da1e5fe01c36d214ddb373"}]},"branch":"refs/heads/master"},"9c5f5c90839a8f1fa7ad62841fe923a16317fca7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-27 01:57:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/92/727592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/727592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/727592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/727592/2"}}},"commit":{"parents":[{"commit":"523d66e8fd5d29be8dbae9aa79c5348d3dce8c64","subject":"Fix CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/523d66e8fd5d29be8dbae9aa79c5348d3dce8c64"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:23:48.000000000","tz":-300},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 01:57:23.000000000","tz":0},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I6a80bd28d4bce6140ba5c3fabe7199301ac026cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c5f5c90839a8f1fa7ad62841fe923a16317fca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c5f5c90839a8f1fa7ad62841fe923a16317fca7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
